)]}'
{"id":"openstack%2Fpuppet-placement~958054","triplet_id":"openstack%2Fpuppet-placement~master~I1b489fd4950d9d31400b0c9c3eb5c63d82257579","project":"openstack/puppet-placement","branch":"master","topic":"wsgi-script","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-23 02:36:51.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I1b489fd4950d9d31400b0c9c3eb5c63d82257579","subject":"Replace wsgi script","status":"MERGED","created":"2025-08-20 11:09:26.000000000","updated":"2025-08-23 05:18:35.000000000","submitted":"2025-08-23 05:18:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958054-wsgi-script","meta_rev_id":"e7a04b489006109f00cc9814272b9ac6d7d218f8","_number":958054,"virtual_id_number":958054,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-23 05:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2025-08-20 14:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2025-08-22 14:07:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2025-08-22 14:07:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-20 12:43:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-20 14:28:23.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-08-22 14:07:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"aec9e5bf85d8bf76a616b3431f6503c40e6c523a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-20 11:09:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd1589a9a623795427dc028ce280a11c18ebb37b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 12:43:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d36adf113d5241e9a9b18a341eae5313\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/743b1e8ecdb848de8cd07e3d427bf605 : SUCCESS in 2m 34s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16b3cd5e4f2244c9a63aab7476ac15cf : SUCCESS in 2m 11s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/11f0f2fd01cc4796935f02ce737bad31 : SUCCESS in 7m 26s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ca95e21ea8c494283811c3afb166490 : SUCCESS in 5m 50s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8e2aaf4f73146b6b44e99c7d45e8124 : SUCCESS in 4m 34s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5539bc3745104f628ba764e49e7248ab : SUCCESS in 32m 19s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57dc87ddf76042f6b1a1a92cd0ab9ee6 : SUCCESS in 33m 30s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7392e04a067449139f63b469a339db11 : FAILURE in 13m 38s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/19e63801c13a4506900d947114527c1b : SUCCESS in 24m 13s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47478c93092c4db5a75ec4dd645734cc : FAILURE in 49m 26s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a459eaa163dc4afeb7d83b31eaaf6826 : SUCCESS in 1h 09m 49s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3afc919417d45878776d860a657dfcf : SUCCESS in 1h 28m 43s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4315aa27de2d4f858e1fb462920e1d83 : SUCCESS in 1h 09m 57s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea22bcdd7ee244be88676a4e3b536501 : SUCCESS in 1h 01m 40s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0aef796f20f6476e889932472ad0a189 : SUCCESS in 43m 09s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ed2c1c5a12d4c12b6161929483fb6e4 : SUCCESS in 50m 35s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7eb0b4ef12e342b2b0c7a45efd5c0c7b : SUCCESS in 1h 25m 22s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f66f316a98514af3a3857ae11cd0311f : SUCCESS in 1h 04m 49s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/db26cbc27feb42bc99efd2244c98b510 : SUCCESS in 1h 00m 25s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7cc45d974fb40f18a88f29506832293 : SUCCESS in 59m 35s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b70e392662a457cb21c0fe6d0ee9253 : SUCCESS in 18m 42s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1c717d60f7349da94f045b3b339606d : SUCCESS in 11m 56s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/053180b7b8c144c89b2f75e895ba7292 : SUCCESS in 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0bb8234a3c942c30a4d7e9ecc8e3419e2d1ea5b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-20 14:28:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1c28f74d6c9a305b9ceeaa9b2c74c820f8f53b35","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-08-22 14:07:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"569d656d87a298e4d1e4d0560bbbc294459acd16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 14:08:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3057c328a39319aafcd695649db91a77057e0381","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 14:43:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1962bf25253748fe919b55b3bfb2a0fe\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9f4a93486eb74c649f2d4f02a396eb5f : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/39a4d799aaae4813ba7e7321eb24e524 : SUCCESS in 2m 27s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/52df2bc4a48e49b49ad02710fa57ba31 : SUCCESS in 4m 10s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66f2143eceae41ad9e18aee50f01b4d9 : SUCCESS in 4m 49s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ccb7ef356f844705bb053e4f439aaa35 : FAILURE in 17m 13s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0c1610452fa4e3a8defcc1fde42b57f : FAILURE in 14m 13s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be8e7f329b424a6f8430ce9428a6c4b6 : FAILURE in 17m 57s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ff75aa694dc48a18eefecb3adebd15b : FAILURE in 13m 28s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2905171248d142a884012abcabd9b7d1 : FAILURE in 17m 28s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/707bfaa99043436ca4ef3d8fec8a70a2 : FAILURE in 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"30ebf806343c8c0d9d67169d5889fe8768b1dd6e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-23 02:36:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de245cc548b2b137f584aaf58307e38449c55b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 03:57:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3050762b756449459dce21a525d0b872\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0662a59a32341b3b86ec4d407cd6dd8 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e673107aaf3840978787203bc8d5f4a3 : SUCCESS in 2m 56s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e71f7b1428d41b18ec44a977527087b : SUCCESS in 4m 23s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0f2187990ece4c2a9be92a24e202a702 : SUCCESS in 4m 35s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bb365ea5877f40b8928aa7591d366829 : SUCCESS in 4m 51s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47cad23c2e5640e3b04afccf3ec4f7bb : SUCCESS in 50m 31s (non-voting)\n- puppet-openstack-integration-8-scenario002-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b33296815b8b428980ba4ea8a7c336fa : SUCCESS in 33m 58s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6879949935241b0baaea425acd12c05 : SUCCESS in 48m 55s (non-voting)\n- puppet-openstack-integration-8-scenario004-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3912dd091d4c41b6bf461bc85c3207c0 : SUCCESS in 48m 45s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e4a08ae89d4408bacd8a51350a02193 : FAILURE in 48m 19s (non-voting)\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/bcd484402d99418a9f3025422fc1b9b0 : SUCCESS in 1h 07m 07s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2954563afb4448318eda5bdd457ca6a7 : SUCCESS in 1h 15m 59s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/df43c1d314c8496fbb472010b5fc4a95 : SUCCESS in 54m 42s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6bac81c3e3c345da9a5f75b2bf96079f : SUCCESS in 46m 10s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1c88370c52548579184f07adac34943 : SUCCESS in 52m 32s\n- openvox-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c07f49b26344d6186f3dd87263ef45e : SUCCESS in 1h 01m 15s (non-voting)\n- openvox-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5330e5f9adc48158fdbf968da6d8c70 : SUCCESS in 47m 06s (non-voting)\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d505ee73983435f981cc5ec78794f59 : SUCCESS in 55m 06s (non-voting)\n- openvox-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5cc39ad596444146ad03c7c9836a69b9 : SUCCESS in 50m 00s (non-voting)\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee1d656239a542e6a0a2d6449590288a : SUCCESS in 28m 30s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b370d5b7e3fd4d4cb20653984f5934ff : SUCCESS in 16m 01s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/27961c45f3aa41f7ad0f5af41004869b : SUCCESS in 9m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a6b768a3126a40deb403490d73fd3083 : SUCCESS in 7m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1956dca966620f9a99255f4d89757f70d906bc21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 03:58:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f4a1ef13f35c2fa0f1756ab8b435650500bb9bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 05:18:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fb34b9798e24ccaa4bef83663322fbd\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d0b224f50224399b48aa79c94a67a88 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eb305b0cb6fc42128614f01fa789340f : SUCCESS in 3m 34s\n- puppet-openstack-unit-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/89995dadf788477295113a675a8f3a50 : SUCCESS in 8m 09s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3b6be013828a4807b57deea830b4f054 : SUCCESS in 4m 11s\n- puppet-openstack-integration-8-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a566baed891e4758874f62f7277a04fa : SUCCESS in 30m 25s\n- puppet-openstack-integration-8-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ef743860a7440b1a251302cab5ad220 : SUCCESS in 1h 16m 39s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/455fe0ba3bf54c36a506519b2293a452 : SUCCESS in 59m 28s\n- puppet-openstack-integration-8-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/70d86510ee38441c90de0e27f6c952d4 : SUCCESS in 38m 23s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8d1b0a463d74cc3839c673f5b5c8107 : SUCCESS in 56m 42s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/29fcf18a56da4499b2c18acce4fe2652 : SUCCESS in 18m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e7a04b489006109f00cc9814272b9ac6d7d218f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 05:18:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3a16e97f74606822ae7b44897f1800995f02c6f","revisions":{"d3a16e97f74606822ae7b44897f1800995f02c6f":{"kind":"REWORK","_number":1,"created":"2025-08-20 11:09:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/54/958054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/54/958054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/54/958054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/54/958054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/54/958054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/54/958054/1"}}},"commit":{"parents":[{"commit":"ef773c0a8c0d19dbf204a80ed197f9dcfd3ba349","subject":"Merge \"Remove puppet 7 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/ef773c0a8c0d19dbf204a80ed197f9dcfd3ba349"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-20 11:09:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-20 11:09:16.000000000","tz":540},"subject":"Replace wsgi script","message":"Replace wsgi script\n\npbr is removing support for wsgi_script due to changes in setuptools.\nUse the wsgi script module python code instead.\n\nNote that this change only touches CentOS code. Ubuntu and Debian will\nbe updated later when we bump the release used in our testing.\n\nChange-Id: I1b489fd4950d9d31400b0c9c3eb5c63d82257579\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/d3a16e97f74606822ae7b44897f1800995f02c6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/d3a16e97f74606822ae7b44897f1800995f02c6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
