)]}'
{"id":"openstack%2Fpuppet-qdr~665812","triplet_id":"openstack%2Fpuppet-qdr~master~I2d1502be16de6bb3c1727c514284372aa5e865e6","project":"openstack/puppet-qdr","branch":"master","hashtags":[],"change_id":"I2d1502be16de6bb3c1727c514284372aa5e865e6","subject":"assert_private() params class","status":"ABANDONED","created":"2019-06-17 23:34:03.000000000","updated":"2020-09-18 22:57:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ccc17165a0b3d16302d27bcc6359608a5d194421","_number":665812,"virtual_id_number":665812,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-17 23:57:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 23:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d31c51135525c93d79614c92dea0de44cdb42d6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:34:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88b14f335d293cf19723c4d7eef15e161dcb61fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 23:57:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/12/665812/1/check/puppet-openstack-lint-ubuntu-bionic/22d6d75/ : SUCCESS in 4m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/12/665812/1/check/puppet-openstack-syntax-5-ubuntu-bionic/921f245/ : SUCCESS in 4m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/12/665812/1/check/puppet-openstack-syntax-6-ubuntu-bionic/a3fcf32/ : SUCCESS in 5m 29s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/12/665812/1/check/puppet-openstack-unit-5.5-centos-7/6e572ff/ : FAILURE in 8m 53s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/12/665812/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/ac593fd/ : FAILURE in 7m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/12/665812/1/check/puppet-openstack-unit-latest-ubuntu-bionic/904e537/ : FAILURE in 7m 26s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/12/665812/1/check/puppet-openstack-beaker-centos-7/53471a5/ : SUCCESS in 8m 14s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/12/665812/1/check/puppet-openstack-beaker-ubuntu-bionic/589dae7/ : SUCCESS in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccc17165a0b3d16302d27bcc6359608a5d194421","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-18 22:57:26.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3340c90e24b585366f54d20951e377466e8b7452","revisions":{"3340c90e24b585366f54d20951e377466e8b7452":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:34:03.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/12/665812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/12/665812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/12/665812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/12/665812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/12/665812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/12/665812/1"}}},"commit":{"parents":[{"commit":"f2c27933ad3ab411363b27765537f2a4c1d4a2d0","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/f2c27933ad3ab411363b27765537f2a4c1d4a2d0"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:33:53.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:33:53.000000000","tz":120},"subject":"assert_private() params class","message":"assert_private() params class\n\nChange-Id: I2d1502be16de6bb3c1727c514284372aa5e865e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/3340c90e24b585366f54d20951e377466e8b7452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/3340c90e24b585366f54d20951e377466e8b7452"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
