)]}'
{"id":"openstack%2Fpuppet-qdr~676050","triplet_id":"openstack%2Fpuppet-qdr~master~I73d7c9fb04c3deb21640512bbe8ab89e88bb65c3","project":"openstack/puppet-qdr","branch":"master","topic":"fix-openstackdocstheme","hashtags":[],"change_id":"I73d7c9fb04c3deb21640512bbe8ab89e88bb65c3","subject":"Bump the openstackdocstheme extension to 1.20","status":"MERGED","created":"2019-08-13 03:30:11.000000000","updated":"2019-08-13 06:10:18.000000000","submitted":"2019-08-13 06:10:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676050-1565676618613-3f7c88c0","meta_rev_id":"3d99e138a26cf1b1a6705b1aa0eec36f1e47ad14","_number":676050,"virtual_id_number":676050,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-08-13 06:10:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-08-13 06:01:35.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-08-13 06:01:35.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 06:10:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ffb0dfe0f84506bb39ac836165ab7b97bd1589bd","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-13 03:30:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22fd5b307afa873d2677e7bdf24ed2a556200fd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 03:33:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb2cc08ad92843eead1dfcced7a0bbe2 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":1},{"id":"869c658ea038ebd7f8ce2dbec49b865789b8646c","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-13 06:01:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"197bc17b625aee4c4de433d5be544418e57dbbf0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 06:01:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf374a33b72051ae2615fc9172680766f0de384c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 06:10:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d3115ce1e150b2cbe30e35ea39d39a42e589a3d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 06:10:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb4b20ba106245dba8ed4713be841ed4 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a36b4b3e1135bc594bcb11912b869bc2bea6aa5b","revisions":{"a36b4b3e1135bc594bcb11912b869bc2bea6aa5b":{"kind":"REWORK","_number":1,"created":"2019-08-13 03:30:11.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/50/676050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/50/676050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/50/676050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/50/676050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/50/676050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/50/676050/1"}}},"commit":{"parents":[{"commit":"3fc3c04ba480f4e3ac03887afc7e979bd3ee3e98","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/3fc3c04ba480f4e3ac03887afc7e979bd3ee3e98"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-13 03:29:07.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-13 03:29:07.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20[1]:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n\n[1]https://docs.openstack.org/openstackdocstheme/latest/#using-the-theme\n\nChange-Id: I73d7c9fb04c3deb21640512bbe8ab89e88bb65c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/a36b4b3e1135bc594bcb11912b869bc2bea6aa5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/a36b4b3e1135bc594bcb11912b869bc2bea6aa5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
