)]}'
{"id":"openstack%2Fpuppet-qdr~707994","triplet_id":"openstack%2Fpuppet-qdr~master~I2d33d5a0427f45370c6cbf862fc8c694afe81214","project":"openstack/puppet-qdr","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I2d33d5a0427f45370c6cbf862fc8c694afe81214","subject":"[ussuri][goal] Cleanup for python 2.7 drop","status":"MERGED","created":"2020-02-16 02:46:04.000000000","updated":"2020-02-18 03:00:08.000000000","submitted":"2020-02-18 02:58:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707994-1581994691542-0cc33de8","meta_rev_id":"bb381fa0c5c7dc4a55db5651c115195acfc01e80","_number":707994,"virtual_id_number":707994,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 02:58:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-02-18 02:40:06.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-02-17 20:33:13.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-02-18 02:40:06.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 20:33:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-18 02:40:06.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-02-18 02:58:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efcd3eb121a19bc3cd2e8ce227ae8843726008f3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-16 02:46:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d6f115b039d7f0c64801893e290b42d1f1ded82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-16 02:55:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7bbcd1f977849abb428e7c1501ed983 : SUCCESS in 5m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0693f30490c4493186c1df40b3268fe9 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7de57aea39d46d09fad79dddc54b9ea : SUCCESS in 5m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/531272f8f8df426e975a13fca31ffed2 : SUCCESS in 8m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/072c5abb43f04c82baac53e37b65505a : SUCCESS in 6m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0739b23110134196bb1a0ea7763415cf : SUCCESS in 7m 01s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/06eef06fe6294d999ddc33146188d750 : SUCCESS in 7m 49s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0cfabb50cebc4f638765b3d5c029878f : SUCCESS in 6m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f23b514e38a14ad7a05507233c810021 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d31fa485892004879430e6b6ae02335bc32197f8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-17 20:33:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"593c031b5fc164eb06ea68392dfd390db49303e9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-02-18 02:40:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2005f09d4d83fabdc9ebd05a9086f51672d18783","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 02:40:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8cca043d067dd97babd55d0abbf1d6420f62c2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 02:58:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1243b3ab68abc9bc76fd8fb63710f96ae312b4b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 02:58:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd9d498a15774d9aa75811ac08b83b2d : SUCCESS in 6m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c49a873db474c2b9e0bdd815804a117 : SUCCESS in 7m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb4dba02d24646aeb566c71daed0ecae : SUCCESS in 6m 17s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8ec8726d0c0d48d6a167410bba856c91 : SUCCESS in 11m 48s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2a4f34fd34746b48b832df6d6d71914 : SUCCESS in 8m 30s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a730c33e5625409a82f4972d3f41edcc : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2928fca7ea0a420ebf1b64b0bb64c7ae : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"4bca5ecd18d55520fa0494d9bec5df8891669c50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 03:00:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8376fcb8736048699133e4280d2be2b4 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee4873d42b6ed4b40de41311d6645039d464c58e","revisions":{"ee4873d42b6ed4b40de41311d6645039d464c58e":{"kind":"REWORK","_number":1,"created":"2020-02-16 02:46:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/94/707994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/94/707994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/94/707994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/94/707994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/94/707994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/94/707994/1"}}},"commit":{"parents":[{"commit":"924781cbe833e8e9f3ea3971e7c618ceb0cf41b7","subject":"Prepare Ussuri M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/924781cbe833e8e9f3ea3971e7c618ceb0cf41b7"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-16 02:45:27.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-16 02:46:03.000000000","tz":-360},"subject":"[ussuri][goal] Cleanup for python 2.7 drop","message":"[ussuri][goal] Cleanup for python 2.7 drop\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\npuppet-* repos have to cleanup the requirement and tox\nfor py2.7 drop.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: I2d33d5a0427f45370c6cbf862fc8c694afe81214\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/ee4873d42b6ed4b40de41311d6645039d464c58e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/ee4873d42b6ed4b40de41311d6645039d464c58e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
