)]}'
{"id":"openstack%2Fpuppet-qdr~724022","triplet_id":"openstack%2Fpuppet-qdr~stable%2Fussuri~Icdde4120bc502c27391831321a96e01b9818aa95","project":"openstack/puppet-qdr","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"Icdde4120bc502c27391831321a96e01b9818aa95","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-28 14:58:11.000000000","updated":"2020-04-29 15:10:08.000000000","submitted":"2020-04-29 15:08:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724022-1588172923439-39316be2","meta_rev_id":"d1907c5a5a9187d4eca986f0b5cd646cf0aa3d15","_number":724022,"virtual_id_number":724022,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"date":"2020-04-29 14:31:04.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 15:08:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-29 01:56:09.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-28 16:47:49.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-29 01:56:09.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 16:47:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-29 01:56:09.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-29 14:31:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-04-29 15:08:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb1190a738925e2227ce75b1a2affb3ed34efef4","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 14:58:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5ee198a45d89c31ee0aed9576ed876408664f57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:07:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea23ada711c64021b2e7f8c5afe3296c : SUCCESS in 7m 05s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cd33097a5c8426e8484ed046c1a9ab2 : SUCCESS in 8m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbcf06c017df41938dfaf1673f98f6d5 : SUCCESS in 6m 27s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2dd3d39cfa5f479983257c9362d36d46 : SUCCESS in 13m 33s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/01c4b80c29d84615947a317af8d35f38 : SUCCESS in 12m 23s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26ae79759bec4ea6bc38a915732d4b31 : SUCCESS in 15m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5faa35b0ee384877a077481ae5d604f7 : SUCCESS in 10m 54s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/aedaefb7a4b747ec9b509e3b85e08e29 : SUCCESS in 12m 15s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b845f9d7c5f246b4ba93c00d1432ddf3 : SUCCESS in 12m 02s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23e34c4ff5914f58baf1f65362e736de : SUCCESS in 12m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/656651efd03748aa87ad7ee332dec358 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":1},{"id":"1211981743839fab24f56ef2d3aecad5365c3352","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-28 16:47:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"39a6c560d0e0b35592d313e05ebcaba92219d46e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-29 01:56:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64bbcb9d83391cd4028769e548b836904a19217b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 01:56:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"838fc2b669cd68595f6dbf9645e821450969d904","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 02:12:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c06d4520c6b64ed997763df2a80d8ddb : SUCCESS in 6m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e46752d940d54de6991e2923ea63ffd8 : SUCCESS in 7m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e48d1147dec423793fc589531f24210 : SUCCESS in 6m 20s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e3209c725fbe4f469991264f4e16d30f : SUCCESS in 12m 18s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/d0c10aac831545358c5f3545d44aba1e : SUCCESS in 12m 01s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25897971de614f878c8054f6095fccec : POST_FAILURE in 7m 32s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bb23053b136b4d3ba83418a674cc0dcc : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3e3e6928f09459aac3dee4f55a3b739 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9f8a731b25dfcbf6e388a3f470ed1684cc63a02b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-29 14:16:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15f741eff8f1407fef985a68f7317c5fe6584d5c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 14:16:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"955c989141bac123dbf986a5c548e80f2010bdd7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 14:31:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"61025153e674ab58af9a20d353f647dac8ec7df7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 14:50:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd4ef14762db437eac3fc5a1108cdf67 : SUCCESS in 6m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8b28dea3e1748bc8473dcffa6382ffc : SUCCESS in 6m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/233a56b2f3504986bfc85a42c072205a : SUCCESS in 7m 31s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e8735797e09b48a49d7459e76e8714f4 : SUCCESS in 12m 35s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/1ab893230a054978ad5c90538ce4571b : SUCCESS in 13m 38s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8da1387aa152479dbdf375675b90b924 : SUCCESS in 11m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01a5b6d6d64a46a4a09258a0cf817080 : SUCCESS in 10m 50s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8ff3af5b5b3049c3b65ee7217de9e4a8 : SUCCESS in 13m 06s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/77cba75053844ab2bc279dd9d025e37f : SUCCESS in 14m 04s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/772b9d076fa24ccab528df5856f3095c : SUCCESS in 9m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b8ae4ca83724af0829ab5a4d9884f73 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"56db81d7782dedf7956cc54055561638c03cc108","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 14:50:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ca8da1723322ece2f4ee047ba865e007a2ebfbc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 15:08:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"28b0cc86d81c5329c18694ffbd9f8e385cd171c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 15:08:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9b8093fe7a341259ff71e6416d91232 : SUCCESS in 6m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35a799b0eddb4070abb376fb487ba7b5 : SUCCESS in 5m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d149c4ff4ab941b1ae6c5a5887d7be81 : SUCCESS in 4m 37s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5b9209897fed45efaa64f56f0c627363 : SUCCESS in 15m 18s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/493a2d4e118649059cac0768146b2111 : SUCCESS in 9m 14s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d6a9bea678d4c3897c91eddc161fce8 : SUCCESS in 7m 48s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a2075183b0124ac8848166d440f4c4eb : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/709b1043031542358b0bc775236b06a3 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5afe5fddf0a1e005c716157a084a63325611c587","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 15:10:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b47d010fa75a47e18c4126253dca3207 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73d9d3f20d0f3907ebbbe56335316033af0c4e40","revisions":{"73d9d3f20d0f3907ebbbe56335316033af0c4e40":{"kind":"REWORK","_number":1,"created":"2020-04-28 14:58:11.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/22/724022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/22/724022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/22/724022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/22/724022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/22/724022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/22/724022/1"}}},"commit":{"parents":[{"commit":"1971e81128735dbf296b4cfb63fcc029a7504255","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/1971e81128735dbf296b4cfb63fcc029a7504255"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:58:09.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:58:09.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Icdde4120bc502c27391831321a96e01b9818aa95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/73d9d3f20d0f3907ebbbe56335316033af0c4e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/73d9d3f20d0f3907ebbbe56335316033af0c4e40"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
