)]}'
{"id":"openstack%2Fpuppet-qdr~728245","triplet_id":"openstack%2Fpuppet-qdr~master~Ifedf296a574a041049f205651ec91a06321fb9d7","project":"openstack/puppet-qdr","branch":"master","topic":"v-version-bump","hashtags":[],"change_id":"Ifedf296a574a041049f205651ec91a06321fb9d7","subject":"Bump version for the start of Victoria","status":"MERGED","created":"2020-05-14 18:27:43.000000000","updated":"2020-05-15 23:13:42.000000000","submitted":"2020-05-15 23:13:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728245-1589584422497-0ef704bd","meta_rev_id":"560c1a799cfbecaf38313f686bcfc29722d44828","_number":728245,"virtual_id_number":728245,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 23:13:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-05-15 23:06:37.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-14 18:45:48.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-05-15 23:06:37.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 18:45:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-15 23:06:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-15 23:13:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a243704c2d461c3bf76732ef9d58695971909389","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-14 18:27:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2dfc78322ca1fd7da8950ab5ad784c074286b3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:42:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/0d6cd57549eb46f7969b7a8ff70e8ba6 : SUCCESS in 3m 40s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce5d929778b84968bff607c52859f87b : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"6f839fd29a749f2a33f1fd7ea50a3645df72c20c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 18:45:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a750a6b5e7650ef9e9886b95a77ed68493653f6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-15 23:06:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9f191b46c8243a29dedb1bf4d1267eb60819a117","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:06:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fa74ba0660a5c8065705f63bd56f4a90bed0064","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:13:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9570bc00788dda88377d811e15c58cb252a129f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 23:13:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/b216304dd2ef458a8180a41714c8b72c : SUCCESS in 2m 40s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3eeda7e8fbc405e8dfea25d4bfdc738 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1cd847bcb092aec2d85b64571a361c8181d167f4","revisions":{"1cd847bcb092aec2d85b64571a361c8181d167f4":{"kind":"REWORK","_number":1,"created":"2020-05-14 18:27:43.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/45/728245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/45/728245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/45/728245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/45/728245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/45/728245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/45/728245/1"}}},"commit":{"parents":[{"commit":"c2a25b1a8b3e038faaebf5e0e1342b5733db525a","subject":"Merge \"[ussuri][goal] Add CONTRIBUTING.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/c2a25b1a8b3e038faaebf5e0e1342b5733db525a"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-05-14 18:24:28.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2020-05-14 18:27:42.000000000","tz":-360},"subject":"Bump version for the start of Victoria","message":"Bump version for the start of Victoria\n\nBump the major version for the Victoria release afte Ussuir has\nbeen branched.\n\nChange-Id: Ifedf296a574a041049f205651ec91a06321fb9d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/1cd847bcb092aec2d85b64571a361c8181d167f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/1cd847bcb092aec2d85b64571a361c8181d167f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
