)]}'
{"id":"openstack%2Fpuppet-qdr~774798","triplet_id":"openstack%2Fpuppet-qdr~stable%2Frocky~Ic4d98dfb30f4883c8084de5db367af4db9db4dca","project":"openstack/puppet-qdr","branch":"stable/rocky","topic":"metrics_qdr_sslprofiles","hashtags":[],"change_id":"Ic4d98dfb30f4883c8084de5db367af4db9db4dca","subject":"MetricsQdr: fix minor spacing issue in qdrouterd template","status":"MERGED","created":"2021-02-10 11:46:43.000000000","updated":"2021-02-24 23:33:54.000000000","submitted":"2021-02-24 23:33:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774798-metrics_qdr_sslprofiles","cherry_pick_of_change":774797,"cherry_pick_of_patch_set":1,"meta_rev_id":"fcdbb45c08452bf305c913ba16c3c3f21c135c69","_number":774798,"virtual_id_number":774798,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 23:33:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"recommended":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2021-02-10 13:53:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2021-02-10 13:34:01.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-02-23 17:57:40.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-11 16:47:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-02-23 17:57:40.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 11:46:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"state":"REVIEWER"},{"updated":"2021-02-10 12:02:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-10 13:34:01.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-02-10 13:53:04.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2021-02-23 17:57:40.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e593cc0b409c7085d5e1258c15ba31ac22dadae1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-10 11:46:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"89124e5a9fedb4d81583ebf55d25a9bdb5a21306","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-10 11:47:16.000000000","message":"Topic metrics_qdr_sslprofiles-stable/stein-stable/rocky removed","accounts_in_message":[],"_revision_number":1},{"id":"68364f4645f77f95f48ff1152f00ed0ac9ed23cd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-10 11:47:25.000000000","message":"Topic set to metrics_qdr_sslprofiles","accounts_in_message":[],"_revision_number":1},{"id":"bf95ede270d3b3bb53f34129951922a06f5cabc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 12:02:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/014ff084ca8e4dbbb9a720f40d22b33f : SUCCESS in 4m 09s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/4ce014c1527346019ee197c8cf2a3d71 : SUCCESS in 3m 59s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/5e9a26d4a57a495e87896a2105cc8a31 : SUCCESS in 3m 51s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/7996d3e9d73f41a0b94fa4fcb1d1d411 : SUCCESS in 8m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7d5dae39fe1d40aaa3825680c3c36c83 : SUCCESS in 12m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5e480601e2444ffd975b2047c9c46119 : SUCCESS in 10m 50s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0828572b30164f8abf2432ba40ff44f6 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7c91ee46af77f4035b4c7bb5c9fbe0573ec4aa62","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-02-10 13:34:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ac84fe79d42762f8cbbfa3682e07813c1553adfd","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-02-10 13:53:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cb24dbe008615f0b3561ccb53725b6174e7599bf","author":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"date":"2021-02-11 16:47:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ba33f4694fecda229ccd1bd52ce3763352ed4f71","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-23 17:57:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c789ed7dc95f385c15c971e056874bf44ff70b89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 17:57:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da9f546a99ae603dc40d476e187d5f28c3a57294","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 18:29:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/a333e679dd2147a68c11890feba7d26a : SUCCESS in 4m 22s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/c035112bd44d482495a555e5824bfd94 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/0557756d34274604a350bee0db297040 : SUCCESS in 4m 46s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/5ce954c6945043118dc03f684f495172 : FAILURE in 7m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b782565b4eef4fa2b575b0c3db7c6488 : FAILURE in 9m 16s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/19edcc37e166498cbded8c01700c8ee0 : SUCCESS in 8m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c92ed0ee6ead4a6197134f5d0a998d13 : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b86a8186cd1aa22a47d9374eb434a43247d55aeb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-24 22:58:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"53cdca59126d9ccd4a63a14ba95d798309cbd1ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 23:15:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/f1ed9d6d188a42bf87722c8a37b3829d : SUCCESS in 4m 43s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/ce96b4266a7b4a0db59621a72d78cd45 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/c6fb62e5812a42bf9daac1c8a4c91e3e : SUCCESS in 5m 02s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/1979915163ca4b848dc088488a5ee841 : SUCCESS in 9m 02s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1a248907b9784872b46e7904a0a102e0 : SUCCESS in 8m 58s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/58fd859ccb96436189079abed6044ef3 : SUCCESS in 9m 13s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d168894f94b48378e257dcfe53548a0 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":1},{"id":"945e490d466722a2623c301254596a97b9d680b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 23:15:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc06e821ead329753dd1c21017d6d8af97652890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 23:33:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/167a941ac0ab4892853a8294c9d7e3f9 : SUCCESS in 4m 08s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/66a2e6b75549499e8e7e41deba8614e0 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/d77a9270738a4168ad43cb1bbe269a2c : SUCCESS in 4m 49s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/87c44847f28f41e694d55c658296e77b : SUCCESS in 10m 27s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c959e8611d324fd1bd10600ea9ad7d89 : SUCCESS in 9m 31s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e4c21c584efb4fe49283980e3c90a7f7 : SUCCESS in 8m 36s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fab3a697e4d49d3a5233b180f8f43e8 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":1},{"id":"fcdbb45c08452bf305c913ba16c3c3f21c135c69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 23:33:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67b99f57adad846c19580dbacac77852b18ff3b8","revisions":{"67b99f57adad846c19580dbacac77852b18ff3b8":{"kind":"REWORK","_number":1,"created":"2021-02-10 11:46:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/98/774798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/98/774798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/98/774798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/98/774798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/98/774798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/98/774798/1"}}},"commit":{"parents":[{"commit":"4d9494c39bc8925ce698a9ef8e07abcf7f86c0cc","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/4d9494c39bc8925ce698a9ef8e07abcf7f86c0cc"}]}],"author":{"name":"Leif","email":"lmadsen@redhat.com","date":"2019-04-26 04:35:01.000000000","tz":-240},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-10 11:46:43.000000000","tz":0},"subject":"MetricsQdr: fix minor spacing issue in qdrouterd template","message":"MetricsQdr: fix minor spacing issue in qdrouterd template\n\nChange-Id: Ic4d98dfb30f4883c8084de5db367af4db9db4dca\n(cherry picked from commit 43522c579f396d2435ce0576b2d0e6bce088b03c)\n(cherry picked from commit 9f043cda37d0cb7d21d6a514ee037970fe4f1e19)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/67b99f57adad846c19580dbacac77852b18ff3b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/67b99f57adad846c19580dbacac77852b18ff3b8"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
