)]}'
{"id":"openstack%2Fpuppet-qdr~777387","triplet_id":"openstack%2Fpuppet-qdr~stable%2Fqueens~Ie72494c7525a2ea361483eadad64f2aba36044f3","project":"openstack/puppet-qdr","branch":"stable/queens","topic":"dnm","hashtags":[],"change_id":"Ie72494c7525a2ea361483eadad64f2aba36044f3","subject":"DNM: Testing gate","status":"ABANDONED","created":"2021-02-24 14:31:07.000000000","updated":"2021-02-27 14:01:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"595e70e8f6a6239e4c78d1dc33a56b25044a3d4c","_number":777387,"virtual_id_number":777387,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-24 14:54:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 14:54:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2796deba53d4f48f49c47538c3bd4da900095190","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-24 14:31:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb68a91f2a077bbe0bc795c48ba45394fd162a14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 14:54:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/90b7f95f7cc74c6c91d4d4a09dd7b62b : SUCCESS in 5m 49s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/0093c548a7484ce69bddf5c62aedc929 : SUCCESS in 5m 12s\n- puppet-openstack-unit-4.8-centos-7 https://zuul.opendev.org/t/openstack/build/4c63e0f352aa4820a1f97580c23f327e : FAILURE in 11m 08s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ff8d6c6673b24ebb8bb91077ce3a7826 : FAILURE in 8m 01s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/cfe58f0345e64e26bc1e76ed881a3a47 : SUCCESS in 8m 20s\n- puppet-openstack-beaker-xenial https://zuul.opendev.org/t/openstack/build/d25052c2909c46e2b7c85b8984640504 : SUCCESS in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"595e70e8f6a6239e4c78d1dc33a56b25044a3d4c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-27 14:01:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1521a597331a6ed59033f3185dfbd7eb2f64e69e","revisions":{"1521a597331a6ed59033f3185dfbd7eb2f64e69e":{"kind":"REWORK","_number":1,"created":"2021-02-24 14:31:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/777387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/87/777387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/87/777387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/87/777387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/87/777387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/87/777387/1"}}},"commit":{"parents":[{"commit":"939f7a60300d869a45474e61ef627b9d9ef187f8","subject":"Remove PE requirement from metadata.json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/939f7a60300d869a45474e61ef627b9d9ef187f8"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-24 14:30:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-02-24 14:30:39.000000000","tz":540},"subject":"DNM: Testing gate","message":"DNM: Testing gate\n\nDepends-on: https://review.opendev.org/777126\nChange-Id: Ie72494c7525a2ea361483eadad64f2aba36044f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/1521a597331a6ed59033f3185dfbd7eb2f64e69e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/1521a597331a6ed59033f3185dfbd7eb2f64e69e"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
