)]}'
{"id":"openstack%2Fpuppet-qdr~777404","triplet_id":"openstack%2Fpuppet-qdr~master~I28b76ae92c5a9ca773d0297483bd1f550b094e41","project":"openstack/puppet-qdr","branch":"master","topic":"qdr-ansible","hashtags":[],"change_id":"I28b76ae92c5a9ca773d0297483bd1f550b094e41","subject":"Adjust yes/no to be booleans","status":"MERGED","created":"2021-02-24 15:21:10.000000000","updated":"2021-03-22 15:17:14.000000000","submitted":"2021-03-22 15:17:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"777404-qdr-ansible","meta_rev_id":"bd5899d171363236fdbe8ba1d360a07e38b96b02","_number":777404,"virtual_id_number":777404,"owner":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 15:17:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":1,"date":"2021-03-05 22:43:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2021-03-22 15:00:04.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-03-05 17:10:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-03-08 15:24:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2021-03-22 15:00:04.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"}],"reviewers":{"REVIEWER":[{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 15:27:53.000000000","updated_by":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-02-24 15:27:53.000000000","updated_by":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-02-24 15:37:35.000000000","updated_by":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"reviewer":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"state":"REVIEWER"},{"updated":"2021-02-24 15:50:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 23:01:26.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2021-02-24 23:29:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-03-05 22:43:41.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"}],"messages":[{"id":"e5481088ed61e8832faa73788f0f26fdff5e0077","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-02-24 15:21:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0413a143237489d02eb7e07f1b0057683f4f78ba","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-02-24 15:27:53.000000000","message":"Patch Set 1:\n\nHopefully the last patch related to these booleans. I\u0027ve re-tested this in a tripleo-standalone environment (including the changes to types and defaults in the deployment on my THT branch) and it all works fine together.","accounts_in_message":[],"_revision_number":1},{"id":"09f626711985dd55e115c099d7e5cf4dea826d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 15:50:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/817d1218121948a4a402bbf757d787a5 : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2ff42d36b9440ccbbbfb284e9580e88 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/18b290fcf67e4b7a86b7fcd3150fd4c9 : FAILURE in 7m 11s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/2eb6b8673fff4079b9e432f7104dc49e : FAILURE in 7m 59s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7478e40ed7254223b142ecc846c45de7 : FAILURE in 6m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83b19131006f441a87e5a5942cbea14f : FAILURE in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fc0308bd4ca1e507e1b6059aaaab16037db0590","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-02-24 16:34:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c3a1242b7527f4cf2c499e5aaf13d926388ea2fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 16:50:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d85c2285631040a0a394fea51dfc94be : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a376b5f5040e4e32aa1f9c7f2c185252 : SUCCESS in 4m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/13be3282948043a795ab78af810480ff : SUCCESS in 9m 34s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0bffe153e8b4338b3eba15cc4a3e7ce : SUCCESS in 9m 31s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da24074c272340e9bf999fad3608fbd0 : SUCCESS in 6m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c5111eaf42f463eb501923237a418db : SUCCESS in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"759fdd7fcfb5fddb7e16e1f5cfb7260f8173c1c4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-24 16:51:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"62ddbf6b1411177abc477a4498e0036c10b73d2c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-24 23:01:26.000000000","message":"Patch Set 2:\n\nCould you explain why we need this change ?\nThe change is a bit concerning because it implements a breaking change in interface.","accounts_in_message":[],"_revision_number":2},{"id":"af859d8b7b2d84e8f668b0439105f3a7f6aa24ff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-24 23:29:50.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"14ff88c3518e0a84975df6fccfd3a8b2f39825dc","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-02-25 14:37:45.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Could you explain why we need this change ?\n\nSo the code conforms to standards and is less confusing to use.\n\n\u003e The change is a bit concerning because it implements a breaking change in interface.\n\nAgreed.\n\n\u003e\n\nHi Takashi, thanks for the question! I\u0027ve been working on a re-implementation of the QDR deployment in Ansible, and discovered this abuse of truthy strings from the automated lint test in the Tripleo Heat Templates repo[1]. Kevin Carter also commented on it and I followed up with him to discuss the situation. These changes are a result of that discussion, which basically boils down to: Use true/false booleans where possible, as long as possible, before translating to yes/no strings. I\u0027ve tested that the underlying service can accept either, so we now carry the booleans intact the whole way without converting to strings. This makes the code more consistent, standards compliant, and more intuitive to use and develop.\n\nHope that helps!\n\n[1] https://review.opendev.org/c/openstack/tripleo-ansible/+/772920/1/tripleo_ansible/roles/tripleo_metrics_qdr/vars/main.yml#35","accounts_in_message":[],"_revision_number":2},{"id":"5f9bd6ad652f591378e08c3a5acef8b819e8e07d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-25 14:42:55.000000000","message":"Patch Set 2:\n\nI think the breaking concern could be alleviated by adding logic to handle yes/no and convert to booleans.","accounts_in_message":[],"_revision_number":2},{"id":"1d4c89c644bf5cf972694f0827b51ba7a3da047c","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-02-25 14:51:36.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e Could you explain why we need this change ?\n\u003e \n\u003e So the code conforms to standards and is less confusing to use.\n\u003e \n\u003e \u003e The change is a bit concerning because it implements a breaking change in interface.\n\u003e \n\u003e Agreed.\n\u003e \n\u003e \u003e\n\u003e \n\u003e Hi Takashi, thanks for the question! I\u0027ve been working on a re-implementation of the QDR deployment in Ansible, and discovered this abuse of truthy strings from the automated lint test in the Tripleo Heat Templates repo[1]. Kevin Carter also commented on it and I followed up with him to discuss the situation. These changes are a result of that discussion, which basically boils down to: Use true/false booleans where possible, as long as possible, before translating to yes/no strings. I\u0027ve tested that the underlying service can accept either, so we now carry the booleans intact the whole way without converting to strings. This makes the code more consistent, standards compliant, and more intuitive to use and develop.\n\u003e \n\u003e Hope that helps!\n\u003e \n\u003e [1] https://review.opendev.org/c/openstack/tripleo-ansible/+/772920/1/tripleo_ansible/roles/tripleo_metrics_qdr/vars/main.yml#35\n\nMinor correction - the original lint tests about truthiness were in the tripleo_ansible repo. While addressing them for the new ansible code, I got errors from the lint tests in THT saying that the legacy code must change to match so we don\u0027t have two THT vars with different types. After fixing THAT it became clear this code would need to adjust as well.","accounts_in_message":[],"_revision_number":2},{"id":"ac46a975cbcc8761fc261538cd2a893b538d605a","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-02-25 14:52:55.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e I think the breaking concern could be alleviated by adding logic to handle yes/no and convert to booleans.\n\nSo the THT variable would go back to being a string?","accounts_in_message":[],"_revision_number":2},{"id":"0121e812e44b51db588ef8ff40fe75233a38a233","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-25 14:59:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c77aa9e87ae4e65c2e54ef6e8b723cfa0d20f4c9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-25 15:07:05.000000000","message":"Patch Set 2:\n\nor another option might be to add a logic  into tripleo-heat-templates so that it passes yes/no to puppet hieradata based on boolean value.","accounts_in_message":[],"_revision_number":2},{"id":"2c815819d66bdde883ddc75e2a6789f9e7cadada","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-25 15:15:35.000000000","message":"Patch Set 2:\n\nwell in THT we expect it to match what is needed in puppet-qdr. So backwards compatibility needs to be handled here and the we\u0027d update THT to use booleans instead","accounts_in_message":[],"_revision_number":2},{"id":"8788a58061a97ed77f12975a943557657beb0279","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-02-25 19:15:28.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e Patch Set 2:\n\u003e \n\u003e well in THT we expect it to match what is needed in puppet-qdr. So backwards compatibility needs to be handled here and the we\u0027d update THT to use booleans instead\n\nOkay, I\u0027ll add the backwards compatible translation here, the THT has already been adjusted to use booleans. Thanks for the assistance, guys.","accounts_in_message":[],"_revision_number":2},{"id":"d494d458c706e479fbb26cb47692a872a45c232a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-03-01 18:43:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3a87ee04a78d24b3ca7528a3c7c78b23990e426b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 18:56:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c514b0f240ca408496a1a4f10e8a0c0b : FAILURE in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3befc55197844e5a23d975bd8ce2016 : FAILURE in 4m 17s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e5e118f32d9a4b95b0901fd374e807e6 : FAILURE in 7m 18s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/271c87e9ddb94acea9086ab43c61d6c8 : FAILURE in 7m 45s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a18e5f59c6614c4e92456067da0ab0a7 : FAILURE in 5m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c5de5c91ee44ba092032887f76624f3 : FAILURE in 6m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3e9ee2b65938ca406e9b0a34b154831668496415","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-03-01 19:20:36.000000000","message":"Patch Set 3:\n\nTests failed with \"An error occurred while installing http-parser (1.2.3), and Bundler cannot\ncontinue.\" which I think has nothing at all to do with my changes.\n\nSpec tests are passing locally:\n\n[csibbitt@csibbitt puppet-qdr]$ git reflog | head -1\n43f8084 HEAD@{0}: reset: moving to 43f80842c7344f10624adf5c614951bdef3adeff\n\n[csibbitt@csibbitt puppet-qdr]$ git status\nOn branch qdr-ansible\nnothing to commit, working tree clean\n\n[csibbitt@csibbitt puppet-qdr]$ rake spec\nrm -rf openstack/puppet-openstack-integration\ngit clone https://opendev.org/openstack/puppet-openstack-integration openstack/puppet-openstack-integration\nCloning into \u0027openstack/puppet-openstack-integration\u0027...\n[...]\nFinished in 4.01 seconds (files took 2.77 seconds to load)\n59 examples, 0 failures\n[...]\n\n[csibbitt@csibbitt puppet-qdr]$ echo $?\n0","accounts_in_message":[],"_revision_number":3},{"id":"fc58ae8280cdd8aad1ebe633f50ecc6cbe0b16dc","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-03-01 19:47:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"24581741783df4ca8835a367c836662f1702beb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 20:02:52.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3864f475c1994410a346c45c3f8a4db4 : FAILURE in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0f6a173de3347c59dbdc9679bd62092 : FAILURE in 3m 50s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e507be07d9cb4947a2a9a16a163ba080 : FAILURE in 8m 36s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/b02e9f1ad3e9498a84c4b86d4b4500ff : FAILURE in 9m 29s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ede724e235c40e5b14d0b86a1acd82b : FAILURE in 7m 52s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/201e8df42b7d4b51ba62a7c67ad7ac73 : FAILURE in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"25bafcd02a47c231e60f1e773e1f04602d42f81b","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-03-02 17:22:17.000000000","message":"Patch Set 3:\n\nCI failures due to https://bugs.launchpad.net/puppet-openstack-integration/+bug/1917457","accounts_in_message":[],"_revision_number":3},{"id":"afdc0b9df3f00bf0a3c0ab953635f48ed09cf826","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-03-05 14:55:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"132bb279ca13a21858567b44e2492704fcfd562e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 15:15:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fb32885182b6464db29f691621d2288a : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d22de01adac45578a7c0ed0b9ecf7e6 : SUCCESS in 4m 24s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c8de03d668dc43dca8a8bb7b9c8c6daf : SUCCESS in 8m 09s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/723a53f565be4c8188201dd7ee0e3a4e : SUCCESS in 8m 35s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9acf5302b5364ebea398510ac55d89fb : SUCCESS in 7m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed261d290435472f91dc0f9a412f0472 : SUCCESS in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"14203816042ea125929bc32bdc58a85479ee7c75","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-05 17:10:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9da0cd09e83836ca8aa3de19e957fd273f3264db","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-03-05 22:43:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"c9d68ec4c3d9984fc0e981df640e7a7c54bb7f46","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-03-08 15:24:36.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5a4ec7e305e1ce5cb2b0cefa0306b8d2c891d12d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-12 14:38:59.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nThank you for updating the patch. The current version looks good. I\u0027ll put +A once I confirm that there are no plan to backport this.\n\nIf you want to backport this to stable branches then we should fix the current implementation so that this change doesn\u0027t trigger change of qdrouterd.conf without any change in puppet parameters. In general we should avoid change in conf files because that triggers restart of services.\nPlease find my subsequent commit which addresses that concern (and my comment inline)","accounts_in_message":[],"_revision_number":3},{"id":"83897cdca967132ce72990a534166c3b810aa878","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-03-22 14:55:30.000000000","message":"Patch Set 3:\n\nThanks, Takashi. There are no plans to backport this change or the related THT changes to older versions.\n\nThanks for the example showing the deprecation; I\u0027d be happy to implement that on the remaining variables in a subsequent patch, I agree that it\u0027s a good idea.","accounts_in_message":[],"_revision_number":3},{"id":"0132a14e53010ef0117d6db6886ab9fe5834fb29","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-22 15:00:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nIf there is no plan to backport this then I\u0027m ok with the current implementation.\nPlease propose a follow up patch to deprecate usage of yes/no (or let me know if you want me to submit the same)","accounts_in_message":[],"_revision_number":3},{"id":"57d24d938256edc0533ccb5973005e7472b49bc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 15:00:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"627e934f4a57aef3ce1ee21889ca54731b677b7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 15:17:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1dec16a3d6fe4df7b32c6c0fe60f7f97 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad846eb4d08644589491dcbe21c9c3bd : SUCCESS in 6m 17s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/2fa7a423851442f59aa667be1111e8cd : SUCCESS in 9m 34s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23002faaf2494a29b7362c2374638a44 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":3},{"id":"bd5899d171363236fdbe8ba1d360a07e38b96b02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 15:17:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"43f80842c7344f10624adf5c614951bdef3adeff","revisions":{"1df825f129f9fb915f57489b900e19fc49df1805":{"kind":"REWORK","_number":1,"created":"2021-02-24 15:21:10.000000000","uploader":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"ref":"refs/changes/04/777404/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/04/777404/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/1"}}},"commit":{"parents":[{"commit":"5c6da3e066d53a9fa73537c54a1ad6c123e7589c","subject":"Prepare Wallaby M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/5c6da3e066d53a9fa73537c54a1ad6c123e7589c"}]}],"author":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-02-24 15:14:59.000000000","tz":-300},"committer":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-02-24 15:21:08.000000000","tz":-300},"subject":"Adjust yes/no to be booleans","message":"Adjust yes/no to be booleans\n\nChange-Id: I28b76ae92c5a9ca773d0297483bd1f550b094e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/1df825f129f9fb915f57489b900e19fc49df1805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/1df825f129f9fb915f57489b900e19fc49df1805"}]},"branch":"refs/heads/master"},"226c6ba7350a0d22e10d7a418f4f9eaf44eb746e":{"kind":"REWORK","_number":2,"created":"2021-02-24 16:34:51.000000000","uploader":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"ref":"refs/changes/04/777404/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/04/777404/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/2"}}},"commit":{"parents":[{"commit":"5c6da3e066d53a9fa73537c54a1ad6c123e7589c","subject":"Prepare Wallaby M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/5c6da3e066d53a9fa73537c54a1ad6c123e7589c"}]}],"author":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-02-24 15:14:59.000000000","tz":-300},"committer":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-02-24 16:33:14.000000000","tz":-300},"subject":"Adjust yes/no to be booleans","message":"Adjust yes/no to be booleans\n\nChange-Id: I28b76ae92c5a9ca773d0297483bd1f550b094e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/226c6ba7350a0d22e10d7a418f4f9eaf44eb746e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/226c6ba7350a0d22e10d7a418f4f9eaf44eb746e"}]},"branch":"refs/heads/master"},"43f80842c7344f10624adf5c614951bdef3adeff":{"kind":"REWORK","_number":3,"created":"2021-03-01 18:43:47.000000000","uploader":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"ref":"refs/changes/04/777404/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/04/777404/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/04/777404/3"}}},"commit":{"parents":[{"commit":"5c6da3e066d53a9fa73537c54a1ad6c123e7589c","subject":"Prepare Wallaby M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/5c6da3e066d53a9fa73537c54a1ad6c123e7589c"}]}],"author":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-02-24 15:14:59.000000000","tz":-300},"committer":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-03-01 18:42:51.000000000","tz":-300},"subject":"Adjust yes/no to be booleans","message":"Adjust yes/no to be booleans\n\nChange-Id: I28b76ae92c5a9ca773d0297483bd1f550b094e41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/43f80842c7344f10624adf5c614951bdef3adeff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/43f80842c7344f10624adf5c614951bdef3adeff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
