)]}'
{"id":"openstack%2Fpuppet-qdr~793285","triplet_id":"openstack%2Fpuppet-qdr~master~Ie6c567b4882c2268f9dc4ea184c9362c16c9b45c","project":"openstack/puppet-qdr","branch":"master","topic":"prep-release-xena-m1","hashtags":[],"change_id":"Ie6c567b4882c2268f9dc4ea184c9362c16c9b45c","subject":"Prepare Xena M1","status":"MERGED","created":"2021-05-27 07:51:16.000000000","updated":"2021-05-31 12:28:21.000000000","submitted":"2021-05-31 12:28:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793285-prep-release-xena-m1","meta_rev_id":"e7309cd75cc16be8c228dab1de4cfa8dc887be04","_number":793285,"virtual_id_number":793285,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-31 12:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-05-31 12:18:41.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-05-31 12:18:41.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 08:00:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 09:20:06.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"1abc826818e58cf23b5a6c4c5584fc6919e0fdbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-05-27 07:51:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14262dc6d8c744b84f7e4085e55c8e913719d3c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 08:00:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/388a628577884d04993e026ce8e9b1a4 : SUCCESS in 3m 15s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa92a9aa2c69481394bd1321806dc923 : FAILURE in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"05591acd851380c46eaf72605cd2e9dce6b9d87a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-27 09:20:06.000000000","message":"Patch Set 1:\n\nLint error will be fixed by https://review.opendev.org/c/openstack/puppet-qdr/+/793327","accounts_in_message":[],"_revision_number":1},{"id":"ddf7156b447480d68992e6ebcb75015bfb0effbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-05-29 02:45:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"62771546ddff5c3cfef75cf01f91a80529d3d23c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-29 02:50:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/5878567e95154deeb16d604fa45062e9 : SUCCESS in 3m 05s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ac4c86c5e1f4abb90b98bcac790b458 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"63975e88cc5bd018c43623a23014e16240e5b8b0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-05-31 12:18:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e4a26c0c56b3bc8812c7be68910eef9b6117b6b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 12:18:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b6fc877ef57eab39069c72aa97e55acb90e45f82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 12:28:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/2899088e04db4f1d9838fbde9fa3e3ca : SUCCESS in 3m 26s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c9878f2f47642e9a0091e2606cb9517 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"e7309cd75cc16be8c228dab1de4cfa8dc887be04","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 12:28:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ce9178d24d673a637f410c0f51ddb0794c2fcdf9","revisions":{"4aad9fd0a9998beecadb6d4d378e922477a1035d":{"kind":"REWORK","_number":1,"created":"2021-05-27 07:51:16.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/85/793285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/85/793285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/85/793285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/85/793285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/85/793285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/85/793285/1"}}},"commit":{"parents":[{"commit":"fd549cd920140a2e85189c47139da4d4ddb71329","subject":"Bump version for the start of Xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/fd549cd920140a2e85189c47139da4d4ddb71329"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-05-27 07:50:27.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-05-27 07:50:27.000000000","tz":480},"subject":"Prepare Xena M1","message":"Prepare Xena M1\n\nUpdate the version metadata for Xena milestone M1\n\nChange-Id: Ie6c567b4882c2268f9dc4ea184c9362c16c9b45c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/4aad9fd0a9998beecadb6d4d378e922477a1035d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/4aad9fd0a9998beecadb6d4d378e922477a1035d"}]},"branch":"refs/heads/master"},"ce9178d24d673a637f410c0f51ddb0794c2fcdf9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-29 02:45:06.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/85/793285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/85/793285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/85/793285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/85/793285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/85/793285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/85/793285/2"}}},"commit":{"parents":[{"commit":"6f47b4c6a52c44252c67448d8a175682b08bc8fa","subject":"Fix lint errors with the latest lint packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/6f47b4c6a52c44252c67448d8a175682b08bc8fa"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-05-27 07:50:27.000000000","tz":480},"committer":{"name":"zhongshengping","email":"chdzsp@163.com","date":"2021-05-29 02:45:06.000000000","tz":0},"subject":"Prepare Xena M1","message":"Prepare Xena M1\n\nUpdate the version metadata for Xena milestone M1\n\nChange-Id: Ie6c567b4882c2268f9dc4ea184c9362c16c9b45c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/ce9178d24d673a637f410c0f51ddb0794c2fcdf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/ce9178d24d673a637f410c0f51ddb0794c2fcdf9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
