)]}'
{"id":"openstack%2Fpuppet-qdr~794417","triplet_id":"openstack%2Fpuppet-qdr~stable%2Fwallaby~Ib8754c6a53c4fbd984583107c418fbd29c924fc3","project":"openstack/puppet-qdr","branch":"stable/wallaby","topic":"bug/1930403-stable/wallaby","hashtags":[],"change_id":"Ib8754c6a53c4fbd984583107c418fbd29c924fc3","subject":"Set fixture paths for unit tests","status":"MERGED","created":"2021-06-03 06:17:59.000000000","updated":"2021-06-03 07:02:38.000000000","submitted":"2021-06-03 07:02:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794417-bug/1930403-stable/wallaby","cherry_pick_of_change":794121,"cherry_pick_of_patch_set":1,"meta_rev_id":"70a78d65a489f5ee85c378e99dcffd571ebf474f","_number":794417,"virtual_id_number":794417,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-03 07:02:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-03 06:18:14.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-03 06:18:14.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 06:43:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7796743fccf03b4e1f67e29c3c6524f8a2917af6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-03 06:17:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"456ac2ad35e7beb4f02e7fab08d63d18b0be99d8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-03 06:18:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f538c2b76fb66bb57b668da663871ef842c24f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 06:43:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59404d68ecae4a0ea2aeebb0c1168434 : SUCCESS in 5m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bb18c83ad6c4df897e6821961ba030c : SUCCESS in 4m 59s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/af2598a299164eb29034ec75a2255d38 : SUCCESS in 11m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54f9d0025c0345d6b29206e22d31fc2c : SUCCESS in 10m 19s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc7380a63c1d4e3f9905c992cbb87446 : SUCCESS in 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9df5f48ac5db255888e3d27497389ecc22d29731","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 06:44:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9cd9676983e99842b9e18f83416710db7035846","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 07:02:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f383be32ede498087e696d8fc0c3d6f : SUCCESS in 6m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65ec78c9025c4897bb4c76624370d8a6 : SUCCESS in 5m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4227b6f54c5d4b4582acb8bc1e63e726 : SUCCESS in 14m 46s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53d8b3500341409286a5cea2ffff76d5 : SUCCESS in 10m 52s","accounts_in_message":[],"_revision_number":1},{"id":"70a78d65a489f5ee85c378e99dcffd571ebf474f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 07:02:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a575de16680c7e8fce258c633ed75d6f62d29a3","revisions":{"8a575de16680c7e8fce258c633ed75d6f62d29a3":{"kind":"REWORK","_number":1,"created":"2021-06-03 06:17:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/794417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/17/794417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/17/794417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/17/794417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/17/794417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/17/794417/1"}}},"commit":{"parents":[{"commit":"9bdde7b05a75799bed991fa5cd4b0b6e03a0febc","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/9bdde7b05a75799bed991fa5cd4b0b6e03a0febc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-02 05:21:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-03 06:17:59.000000000","tz":0},"subject":"Set fixture paths for unit tests","message":"Set fixture paths for unit tests\n\nThis change defines manifest_dir and module_path expicitly in unit\ntests so that modules installed under fixtures directory is properly\nloaded.\n\nCloses-Bug: #1930403\nChange-Id: Ib8754c6a53c4fbd984583107c418fbd29c924fc3\n(cherry picked from commit fc36b7462980d0b6641c2be0230117795898faf5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/8a575de16680c7e8fce258c633ed75d6f62d29a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/8a575de16680c7e8fce258c633ed75d6f62d29a3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
