)]}'
{"id":"openstack%2Fpuppet-qdr~802271","triplet_id":"openstack%2Fpuppet-qdr~master~I14352371720203a7f1f90a6630d57e67c5be0d36","project":"openstack/puppet-qdr","branch":"master","topic":"prep-release-xena-m2","hashtags":[],"change_id":"I14352371720203a7f1f90a6630d57e67c5be0d36","subject":"Prepare Xena M2","status":"MERGED","created":"2021-07-26 11:04:24.000000000","updated":"2021-07-26 15:57:10.000000000","submitted":"2021-07-26 15:57:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802271-prep-release-xena-m2","meta_rev_id":"6fed3bf68843a5eb5f0b204d0abe9fcc90f5a8eb","_number":802271,"virtual_id_number":802271,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-26 15:57:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-07-26 15:49:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2021-07-26 15:49:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-26 11:10:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-26 15:49:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"08d75eafa76124e9c2e9202f67198a995c9bba9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-07-26 11:04:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"429d9cdfa7900602f295f11086f02e6db532704c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 11:10:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/2a5f8dd77c974f04bd1f773acbcbc8ea : SUCCESS in 3m 36s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/604a88faf20f484aac078844bcca7126 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4de70da2749a3411338d23134556fb69665e995e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-26 15:49:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7dce7bb47f83604de4a0d29c8c603a3b6003698d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 15:49:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8af2edfbb3af82c7655f5ccb66828132a33ab5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 15:57:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/2f8636a05eb141c2a1fa41377f75ef31 : SUCCESS in 3m 17s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d685a0c74ac14630917972cdbaa7f88d : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"6fed3bf68843a5eb5f0b204d0abe9fcc90f5a8eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 15:57:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a134d1032530308b0646f7f56d9a8fe1390d0e51","revisions":{"a134d1032530308b0646f7f56d9a8fe1390d0e51":{"kind":"REWORK","_number":1,"created":"2021-07-26 11:04:24.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/71/802271/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/71/802271/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/71/802271/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/71/802271/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/71/802271/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/71/802271/1"}}},"commit":{"parents":[{"commit":"0d9350d6e37dbb2bcfaffbbacf4c933daf7da65f","subject":"Merge \"Drop Fedora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/0d9350d6e37dbb2bcfaffbbacf4c933daf7da65f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-07-26 11:03:47.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-07-26 11:03:47.000000000","tz":480},"subject":"Prepare Xena M2","message":"Prepare Xena M2\n\nUpdate the version metadata for Xena milestone M2\n\nChange-Id: I14352371720203a7f1f90a6630d57e67c5be0d36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/a134d1032530308b0646f7f56d9a8fe1390d0e51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/a134d1032530308b0646f7f56d9a8fe1390d0e51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
