)]}'
{"id":"openstack%2Fpuppet-qdr~885732","triplet_id":"openstack%2Fpuppet-qdr~master~I534f29d75046a854a54b4c07cfb5757bfd39dd92","project":"openstack/puppet-qdr","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-08-02 20:23:44.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-08-02 08:26:37.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I534f29d75046a854a54b4c07cfb5757bfd39dd92","subject":"Fix wrong default value of qdr::router_worker_threads","status":"MERGED","created":"2023-06-09 08:57:25.000000000","updated":"2023-08-02 20:24:40.000000000","submitted":"2023-08-02 20:23:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885732","meta_rev_id":"8ca1d49c5cb816fe43f3dd629db84c8fd5b322f0","_number":885732,"virtual_id_number":885732,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-02 20:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-08-02 08:26:37.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-08-02 19:11:56.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-08-02 19:11:56.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-09 09:07:57.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-06-09 10:11:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-02 19:11:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"547fbee721190ad5faaaa540617fab6a509be7df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-09 08:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ca1d922dd4d5af854331767224879a4a0f18296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-09 08:59:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0fb375ba21a4f6f8bb1927503e78b1573a44af49","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-06-09 09:07:57.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1fdee80c8f7ee7b680f6daaf1b9039286ba385b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-09 10:11:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/867e4c2ec34f42519134651311e33672\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffc673232b0f4547995ff05f698291cc : SUCCESS in 3m 33s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6424381184e4276bf70779a8150f17c : SUCCESS in 3m 21s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/04dbf22ecab249248af83405feb115ab : SUCCESS in 5m 43s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/f117aaf60f5b4164a5dad8b6c87eddef : FAILURE in 10m 03s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb13dc95a94e491384c7a61cccb94bcf : FAILURE in 7m 22s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/d22bd4f98412456ab018132f451be790 : FAILURE in 8m 45s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d2b88e991f8437293483c61c8203571 : FAILURE in 5m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aaa5d1f54c7f4ffbbae28f80add4e2cb : SUCCESS in 42m 38s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f69a1bdfa4c40ce8ed671fb04deb684 : SUCCESS in 1h 05m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64f06eca806d4fc1b33c75a187a7de34 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c361427a38b098f3bd9017b7f432c125fc9cb347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-13 18:22:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cf8cd2e21cdec614582f1851295bb8061a640e0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-13 19:19:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c2d38782dc441c9983c2e52830115e9\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/948236b80d6f475594d466f7135ca45d : SUCCESS in 8m 44s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/559e7546fa6e41d7a2f56b6d1c5aab2a : SUCCESS in 2m 32s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3955b4e60cf4644b499e654e457fb7b : SUCCESS in 5m 39s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/01b539a215be40be9c3a255bb679c968 : FAILURE in 7m 45s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/150743101e254d35ac5d193a3c0f0d5d : FAILURE in 6m 04s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/5330e924f6f94609b1c813bbd73a27e1 : FAILURE in 8m 07s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25b38403111a4641ac4c2741b7f41aea : FAILURE in 4m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb8a8eb93a564b608a95b5cd5f606124 : FAILURE in 5m 28s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0808d44cbe8f48a59144f2e93b317dc6 : SUCCESS in 53m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c808446b1e84955ac8dfc8ab6629683 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":3},{"id":"781855bfdb4d22bd37967590e933f7edf7dc9c1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-13 03:31:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d5de4066f1e41a6fad006b36d4d8c3fd1c52ab66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-13 03:39:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9a7b3928c61b6d2dce7de456dd68d70776939f5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 04:30:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36e490a7b37e47e8b096b9c8c4cd0b7e\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/614cf52cd3984a6eb14085e895517266 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dee61898fb574e45bcfd5a94399a9301 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d690e940218b4ce0afcf274478432caf : SUCCESS in 6m 02s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/c907ea53ab2f409c93e5b166e5ed49c4 : FAILURE in 7m 57s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/615db07328404268958d135557bbe5fc : FAILURE in 6m 55s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d07c2ac5e784831b741a3fc8d62c8b8 : FAILURE in 10m 46s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b88f376fc73a460a8cebda65485ae8c4 : FAILURE in 8m 04s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5fead1ffa2542759a8984b585f8a3aa : SUCCESS in 43m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e43b2d1c57834dc58e46c305debc3d55 : SUCCESS in 49m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa8e2f8655034d4692e4fd0bccee6d6e : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":5},{"id":"a10874e73336db8eeb27a68719d7db3132c69cbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-13 05:49:29.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6c4654a32273d8a77984fc4dc411221bb5519283","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 06:37:43.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4472340af6544a76b4d9d18164f36bd6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce19e507b44f46768002c37b356f7c9e : SUCCESS in 2m 15s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8434c0aad0a240bf8951803efb089a88 : SUCCESS in 2m 21s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/6381cbad5cf04f818a2c822c4f57f282 : SUCCESS in 5m 31s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0c50071533940bb9715eda554ebaea0 : FAILURE in 8m 58s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0efe6c8d4ce5434da66afa16c50e7082 : FAILURE in 5m 10s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/581936b4b3c241a2912da48c12b71d8c : FAILURE in 8m 44s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44bc6c050fdc46dc9fa97c2668c286c3 : FAILURE in 5m 22s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c12a87a82fc04cdda82e4bbbbe1d4977 : SUCCESS in 42m 30s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c1bc95812c448c58bc5a35cdac0e5f8 : SUCCESS in 46m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38861db80506412d94a24cf5746cc954 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":6},{"id":"3e9cb6425b295bd471542750dabe318c45d660f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-13 06:55:33.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d2c94c5c533de4126779a718f7309a0c6dbdcdc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 07:48:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32a5d391f43d412d858a269e308a9251\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1375bd7a00e04a82acd507987c8c29d2 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cfa42b9474047b38e472a9575061edb : SUCCESS in 2m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/132ec007dbda4b4c82fc68d92e2b3dd1 : SUCCESS in 5m 24s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/36407aadb8884ece9c1235e9244ee5b4 : FAILURE in 8m 02s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bd5635322c145159295a1df14ed4bd6 : FAILURE in 5m 27s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3f9d2cab2cd4233bc83af4d035f7417 : FAILURE in 8m 57s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/700f8a1c5aeb45c594703e052d52009a : FAILURE in 5m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba73d025512a474c8ae0202d15913ae4 : SUCCESS in 43m 07s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a36217a419434c6b92de99f749b327ea : SUCCESS in 51m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93b55105b4164c38a635d4ecfb09ac55 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":7},{"id":"403edd2fb5b7a3411a8e42527c6e632ed91a31a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-13 09:51:17.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"da45d8149505af677559ff83f89f7e532795994b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 11:05:54.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db130de630f94728acd81fd41b74a9f4\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15a0728b575d4528a0d1c3d9df576fc8 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0342320939b4be39cfa9ecdb18b81f5 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e73d3e70fcb41b995ae4fd54e886d47 : SUCCESS in 5m 57s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7cd72fe89b44426be930bb3f2637c56 : FAILURE in 12m 29s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c6a3a2a39084d658049b0fd061cddb6 : FAILURE in 6m 02s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbea582df0e54380b1ca4761d4ca5168 : FAILURE in 7m 42s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/982bb084d0ee4c5d90bbec64c23f5f8f : FAILURE in 6m 06s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a97ada0765964a2daf856af0ab36090c : SUCCESS in 43m 05s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d36b9d163ff544ceb422fcc4f437b06a : SUCCESS in 57m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd2643d8c48e4756bc68bd000f143928 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":8},{"id":"0e534f2dc57204c32529c0532907b4888a629cdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-13 12:49:39.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"70608c8b3522601e9464a1db1582ab1f295bae61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-13 13:56:22.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6613b37fbc6340f3af4186cfaf430fde\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee45ad37659d4cc391c31cf3cde7e5f2 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aeffd98da92545a8b930f6946315fd5b : SUCCESS in 2m 49s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/855aea2dacc14ae5a38594af35a1710c : SUCCESS in 5m 02s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/34e0a8b4417f4b448d7d7e28c0329234 : FAILURE in 9m 12s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d468e7d8b5b456e8596e3904836c9c6 : FAILURE in 6m 12s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f25410fa3894a32a68b1f8ffde64f8f : FAILURE in 8m 42s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bda03ed4ff4c4f71be07ad1d19ef90cb : FAILURE in 5m 37s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/839bc32b18bd42078315528decf0f7c1 : SUCCESS in 49m 11s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f40111570dfb4fe8aea04d6497344415 : SUCCESS in 1h 01m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60ff6cd2bb464e818d43d5936cc37e7c : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":9},{"id":"b375b9e1720b2cba329074c14bb1ee9a4bbbe7c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-07-16 09:32:25.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"c5bf16695e0c3d86ae960ae433a15f58e4a47179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-16 10:33:47.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c46a605cf08b46df9d524a5fdd9e718a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbd07ad3419944b2b9fb590847a4339f : SUCCESS in 2m 11s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46bcf830b5934e0ab20407c78204703e : SUCCESS in 2m 14s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/80f8b160005d45e39976c399810b0b8a : SUCCESS in 4m 30s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/35bc28e5f37e42d8be75962e94bc45e7 : FAILURE in 8m 28s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db10d521730d42f9b5efb5315fb7e036 : FAILURE in 5m 25s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1a632433078478098d65d832c3f1d71 : FAILURE in 6m 53s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad29bf7de39b427296f0a336d15083b0 : FAILURE in 4m 33s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d977b4f01f244a4e91b3785c201fea3e : SUCCESS in 53m 32s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/851399b512da495a8fb854373ef2670d : SUCCESS in 47m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c7a4d0ddd3442e2b95f9c6e90aee00a : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":10},{"id":"2b6a76d35abea86ec2e548eb8f50c20a0228f13a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-01 17:43:12.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"28003d5ca2d848e67e2e9c061cf4b8351ce8e802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 19:05:04.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8ce175c29ab42ae93230aa087471a0a\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4d33e66cbbe4a7cb4bc9f21c5e6d5ff : SUCCESS in 3m 06s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f78e141bf73c4e769c17255c36d7ad65 : SUCCESS in 3m 16s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/d12605e0a9db4ceab80ba5325af92654 : SUCCESS in 5m 38s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0e309fb49e940f5b0225b54b18632c8 : FAILURE in 9m 57s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44141590ede540098896210188ac687e : FAILURE in 7m 17s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/c22e481c8a8a4d9796397f5a05f733ef : FAILURE in 8m 36s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa2771ed928843e0a6e85efa1feb95f7 : FAILURE in 7m 24s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4df9f41208b4b46b22b959ba3687102 : SUCCESS in 52m 27s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/46939b8e9e85441c9ee72771d68bdd08 : SUCCESS in 1h 14m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c02d0d76f0624cab8da2d3411d42d1da : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":11},{"id":"5347a4180dcc6566bad9fe6654b751b583382f16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-02 03:02:17.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"f35ed08543523e4df0c8d820773e31fd855f98d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-02 03:55:27.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"29fb6f3f34755d72b4fa74305caf40876f6f2d7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 04:57:42.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa47e0a0fd9e42a4a734c912270a7901\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3818a156888a4f9bbbdddc5f73e1f647 : SUCCESS in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d6dcd261fed44cdb6a0a94d0eba29aa : SUCCESS in 2m 36s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/a59f36ad0174461095f032af29d32d69 : SUCCESS in 5m 02s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d62f64e613d4aa7abf44e77dbc1efde : FAILURE in 11m 18s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45b66f28205c48a6afc3723a8b42b62d : FAILURE in 5m 47s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/880cd1afa484455a9c01ac0fdfa47ca9 : FAILURE in 8m 48s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4a9e948029e4c61b8d1a2c6f50a33a5 : FAILURE in 6m 25s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bde61fb3676490d952146c6a481eccb : SUCCESS in 47m 20s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2454cb5b95f041e693e74032fb96d13b : SUCCESS in 1h 00m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35deb0a1f2b0419c9d2e5392a02fb881 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":13},{"id":"64e29452b905f631cbf7f0bc4c255dff9bc35301","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-02 05:04:14.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"c80834ad3d1def1167cd4defdccc150f295b783f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 06:13:15.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66d3912693114ee198e734543c714e26\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bf3d82bf0e94dfaa27ba24cd4306a75 : SUCCESS in 3m 17s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86131d6c5e494ffa862469eb959c3ab4 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdef8d08b2904e0ab787d2bff7cffb93 : SUCCESS in 4m 57s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc89511730ee409ebb107697199a7859 : SUCCESS in 8m 24s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37f2b792772b4047a12e4d3fdfad3c53 : SUCCESS in 5m 49s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7cb3c07e164436a8cae7262b6a63f63 : SUCCESS in 8m 18s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6c6076ebd7d14c9685c01a5b87d63db3 : SUCCESS in 5m 54s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/100b7d32398844b3b220d7999d075530 : SUCCESS in 51m 10s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5b24cfa3724482aa622829c8b788fe4 : SUCCESS in 1h 07m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c4a5d118f3a43ec89b92993e7c56394 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":14},{"id":"498266f1c283e56fdea351f8a2d57064400edcd6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-08-02 08:26:37.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"9999317ffd017207e6406da1fb1b2cabe8616d70","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-08-02 19:11:56.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"a7fdc7fbed1765da8d9a715d37f6e248b2cf14bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 19:12:48.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"e3d8a2d3f3dfa3adbf5aba507e95042ce661ba4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 20:23:44.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91f28bbc14904a7faf13afb969c6bcb1\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43015370b65d47c3bbef07ac17df494f : SUCCESS in 2m 25s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4590155e4d7947908f08fc80608684e3 : SUCCESS in 5m 15s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/92e67c2fcff44a7aab9f6ec2a2924b17 : SUCCESS in 9m 18s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e9ddf6541204567812c3abb05c6fbb7 : SUCCESS in 6m 24s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/b98fd17cea454fdbb508ecd664cfc1dd : SUCCESS in 9m 56s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c937ac4a92349b1a513585e9bb8b094 : SUCCESS in 1h 06m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c6ecb149aab4521b0ac71ee8689218a : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":14},{"id":"49789a43b0a8b9a2b2ce86d19b53ccc1363850bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 20:23:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"8ca1d49c5cb816fe43f3dd629db84c8fd5b322f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 20:24:40.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e97eee25f86742a2980f193e41dd6225\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fd8db4a30d64088bd2e2ebe8791cda3 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"b50afaee830a11188126933fed601f80af09737f","revisions":{"21d53ddf20a65021ec24c61a731969910aff5286":{"kind":"REWORK","_number":1,"created":"2023-06-09 08:57:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/1"}}},"commit":{"parents":[{"commit":"9a7a6c3f43b1a65d4cb8178d31915795d7e0d0e1","subject":"Replace Stdlib::Absolutepath by Stdlib::Compat::Absolute_path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/9a7a6c3f43b1a65d4cb8178d31915795d7e0d0e1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/21d53ddf20a65021ec24c61a731969910aff5286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/21d53ddf20a65021ec24c61a731969910aff5286"}]},"branch":"refs/heads/master"},"d34590fcd4c0a8fccd77c3c212d44b49a85c3fd8":{"kind":"REWORK","_number":2,"created":"2023-06-09 08:59:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/2"}}},"commit":{"parents":[{"commit":"9a7a6c3f43b1a65d4cb8178d31915795d7e0d0e1","subject":"Replace Stdlib::Absolutepath by Stdlib::Compat::Absolute_path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/9a7a6c3f43b1a65d4cb8178d31915795d7e0d0e1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:59:41.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d34590fcd4c0a8fccd77c3c212d44b49a85c3fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d34590fcd4c0a8fccd77c3c212d44b49a85c3fd8"}]},"branch":"refs/heads/master"},"c590bd5076354f198b0dc225e633d323501526f2":{"kind":"REWORK","_number":3,"created":"2023-06-13 18:22:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/3"}}},"commit":{"parents":[{"commit":"9a7a6c3f43b1a65d4cb8178d31915795d7e0d0e1","subject":"Replace Stdlib::Absolutepath by Stdlib::Compat::Absolute_path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/9a7a6c3f43b1a65d4cb8178d31915795d7e0d0e1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-13 18:22:10.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/c590bd5076354f198b0dc225e633d323501526f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/c590bd5076354f198b0dc225e633d323501526f2"}]},"branch":"refs/heads/master"},"b68f5c39392f0334036cf72dadb845637319e851":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-07-13 03:31:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/4"}}},"commit":{"parents":[{"commit":"d78a346455cbff2fd7f0dd04296648e9c65b3b6c","subject":"Merge \"Replace deprecated Stdlib::Compat::Integer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d78a346455cbff2fd7f0dd04296648e9c65b3b6c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-13 03:31:48.000000000","tz":0},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/b68f5c39392f0334036cf72dadb845637319e851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/b68f5c39392f0334036cf72dadb845637319e851"}]},"branch":"refs/heads/master","description":"Rebase"},"6967492ef4419c3cb9475990376001bb1ea02d7d":{"kind":"REWORK","_number":5,"created":"2023-07-13 03:39:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/5"}}},"commit":{"parents":[{"commit":"d78a346455cbff2fd7f0dd04296648e9c65b3b6c","subject":"Merge \"Replace deprecated Stdlib::Compat::Integer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d78a346455cbff2fd7f0dd04296648e9c65b3b6c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-13 03:39:02.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/6967492ef4419c3cb9475990376001bb1ea02d7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/6967492ef4419c3cb9475990376001bb1ea02d7d"}]},"branch":"refs/heads/master"},"396ee21116294c30e2a68c2636260f2ca04ea528":{"kind":"REWORK","_number":6,"created":"2023-07-13 05:49:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/6"}}},"commit":{"parents":[{"commit":"d78a346455cbff2fd7f0dd04296648e9c65b3b6c","subject":"Merge \"Replace deprecated Stdlib::Compat::Integer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d78a346455cbff2fd7f0dd04296648e9c65b3b6c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-13 05:49:29.000000000","tz":0},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/396ee21116294c30e2a68c2636260f2ca04ea528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/396ee21116294c30e2a68c2636260f2ca04ea528"}]},"branch":"refs/heads/master"},"3419d4cab73a1e065a30a97f2160dd7200965219":{"kind":"REWORK","_number":7,"created":"2023-07-13 06:55:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/7"}}},"commit":{"parents":[{"commit":"d78a346455cbff2fd7f0dd04296648e9c65b3b6c","subject":"Merge \"Replace deprecated Stdlib::Compat::Integer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d78a346455cbff2fd7f0dd04296648e9c65b3b6c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-13 06:55:24.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/3419d4cab73a1e065a30a97f2160dd7200965219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/3419d4cab73a1e065a30a97f2160dd7200965219"}]},"branch":"refs/heads/master"},"d272bdbe017540a9c16a424d50582a9fc883303d":{"kind":"REWORK","_number":8,"created":"2023-07-13 09:51:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/8"}}},"commit":{"parents":[{"commit":"d8afefaa60b553676dc3f82b4930770a754a4b6c","subject":"Ensure tools packages are installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d8afefaa60b553676dc3f82b4930770a754a4b6c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-13 09:51:06.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d272bdbe017540a9c16a424d50582a9fc883303d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d272bdbe017540a9c16a424d50582a9fc883303d"}]},"branch":"refs/heads/master"},"577136e98cb5c672d9448b11a6457dbf37851913":{"kind":"REWORK","_number":9,"created":"2023-07-13 12:49:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/9"}}},"commit":{"parents":[{"commit":"d8afefaa60b553676dc3f82b4930770a754a4b6c","subject":"Ensure tools packages are installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d8afefaa60b553676dc3f82b4930770a754a4b6c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-13 12:48:52.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/577136e98cb5c672d9448b11a6457dbf37851913"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/577136e98cb5c672d9448b11a6457dbf37851913"}]},"branch":"refs/heads/master"},"34a2468b47187e534447ce60df296b52efb17659":{"kind":"REWORK","_number":10,"created":"2023-07-16 09:32:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/10"}}},"commit":{"parents":[{"commit":"d8afefaa60b553676dc3f82b4930770a754a4b6c","subject":"Ensure tools packages are installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d8afefaa60b553676dc3f82b4930770a754a4b6c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-07-16 09:32:15.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/34a2468b47187e534447ce60df296b52efb17659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/34a2468b47187e534447ce60df296b52efb17659"}]},"branch":"refs/heads/master"},"40280b70b8d7a8bb28c50a76875b8e877d78e149":{"kind":"REWORK","_number":11,"created":"2023-08-01 17:43:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/11"}}},"commit":{"parents":[{"commit":"d8afefaa60b553676dc3f82b4930770a754a4b6c","subject":"Ensure tools packages are installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d8afefaa60b553676dc3f82b4930770a754a4b6c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-01 17:43:01.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/40280b70b8d7a8bb28c50a76875b8e877d78e149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/40280b70b8d7a8bb28c50a76875b8e877d78e149"}]},"branch":"refs/heads/master"},"42f8d3b356402bf427c133aa470551dfbdabb0b3":{"kind":"REWORK","_number":12,"created":"2023-08-02 03:02:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/12"}}},"commit":{"parents":[{"commit":"d8afefaa60b553676dc3f82b4930770a754a4b6c","subject":"Ensure tools packages are installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d8afefaa60b553676dc3f82b4930770a754a4b6c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-02 03:02:09.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/42f8d3b356402bf427c133aa470551dfbdabb0b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/42f8d3b356402bf427c133aa470551dfbdabb0b3"}]},"branch":"refs/heads/master"},"92ffe0ac58f097890c3cee93c58be83fdcf0535a":{"kind":"REWORK","_number":13,"created":"2023-08-02 03:55:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/13"}}},"commit":{"parents":[{"commit":"d8afefaa60b553676dc3f82b4930770a754a4b6c","subject":"Ensure tools packages are installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d8afefaa60b553676dc3f82b4930770a754a4b6c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-02 03:55:18.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/92ffe0ac58f097890c3cee93c58be83fdcf0535a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/92ffe0ac58f097890c3cee93c58be83fdcf0535a"}]},"branch":"refs/heads/master"},"b50afaee830a11188126933fed601f80af09737f":{"kind":"REWORK","_number":14,"created":"2023-08-02 05:04:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/885732/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/32/885732/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/32/885732/14"}}},"commit":{"parents":[{"commit":"d8afefaa60b553676dc3f82b4930770a754a4b6c","subject":"Ensure tools packages are installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/d8afefaa60b553676dc3f82b4930770a754a4b6c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-06-09 08:56:42.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-08-02 05:04:04.000000000","tz":540},"subject":"Fix wrong default value of qdr::router_worker_threads","message":"Fix wrong default value of qdr::router_worker_threads\n\nThe default should match processor counts instead of os_worker fact.\nThis module does not depend on openstacklib so we should not use\nthe fact provided that module.\n\nCloses-Bug: #2023371\nChange-Id: I534f29d75046a854a54b4c07cfb5757bfd39dd92\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/b50afaee830a11188126933fed601f80af09737f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/b50afaee830a11188126933fed601f80af09737f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
