)]}'
{"id":"openstack%2Fpuppet-qdr~907952","triplet_id":"openstack%2Fpuppet-qdr~master~I77f7bc94b8d1b3459c64f44f5d18d5f8c747ae4a","project":"openstack/puppet-qdr","branch":"master","topic":"retire-puppet-qdr","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-22 01:46:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-02-22 01:47:25.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2024-02-22 01:47:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I77f7bc94b8d1b3459c64f44f5d18d5f8c747ae4a","subject":"Retire puppet-qdr: Remove Project Content","status":"MERGED","created":"2024-02-06 07:05:41.000000000","updated":"2024-02-22 01:47:25.000000000","submitted":"2024-02-22 01:47:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907952-retire-puppet-qdr","meta_rev_id":"7cd642dae039270be97b61bf78e107174a590803","_number":907952,"virtual_id_number":907952,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-22 01:47:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-02-22 01:46:47.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-02-22 01:46:47.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 07:10:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-06 23:02:44.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-02-16 12:14:11.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"73634b6dfc5908d219c06f943eebf5887a2f25f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 07:05:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bb3a52b90cb253ad907d9fa26e2679baa2cd5be","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 07:09:32.000000000","message":"Topic set to reqire-puppet-qdr","accounts_in_message":[],"_revision_number":1},{"id":"18fbd0855177eb04ba13f523d4f80b4884fece64","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 07:09:39.000000000","message":"Topic reqire-puppet-qdr removed","accounts_in_message":[],"_revision_number":1},{"id":"ff4bb73ac4986ede1eaced8bd49d9348aff2f081","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 07:09:47.000000000","message":"Topic set to retire-puppet-qdr","accounts_in_message":[],"_revision_number":1},{"id":"6a7fa2f9eea0af33d11503497ae49850bc6877ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 07:10:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d9c9fb65c0041b0870a2f6ea7dea3c1\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/69ab2ba64a47432c8e10c3237ac0fb14 : FAILURE in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a572eec02bbbdd94bd9695141aaf08810e15a4fc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-06 23:02:44.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7005e467f40d277996173a3e7c9f1baf256a9bb4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-02-16 12:14:11.000000000","message":"Patch Set 1: Code-Review+1\n\nI\u0027ve approved the dependency.","accounts_in_message":[],"_revision_number":1},{"id":"b4d6d8a02c7d1a14550943b12136dfae5b8ade65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-22 01:46:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"adf4add2bcf6e53d260e80793dbf07633372ba38","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-22 01:46:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"03264beb30f80fe5c55cd8f32c71168630c89d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 01:47:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe70bfa29dd24f20892bda42548b354e\n\n- noop https://zuul.opendev.org/t/openstack/build/e428cece2e3440f59bb8ba64dba6fba8 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"ccf1cb5c9d383d95106355544f91b52c0a4c1ce5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 01:47:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"083ac5f877d41934362bce9eaed9d855ef47167c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 01:47:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de669d8825a8457f8722431442814a76\n\n- noop https://zuul.opendev.org/t/openstack/build/c987d47e4a464a18a06032b98c474e24 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"7cd642dae039270be97b61bf78e107174a590803","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 01:47:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"31aeb84bae9f82f8550000db1029b9d1c856d051","revisions":{"89a7097f466486fdbafad00e00f00617568af17b":{"kind":"REWORK","_number":1,"created":"2024-02-06 07:05:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/907952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/52/907952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/52/907952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/52/907952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/52/907952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/52/907952/1"}}},"commit":{"parents":[{"commit":"1bc79f51d586abe5bac7b3d9a68bad163a5529be","subject":"Bump supported Debian version to 12 (Bookworm)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/1bc79f51d586abe5bac7b3d9a68bad163a5529be"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 07:04:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 07:04:56.000000000","tz":540},"subject":"Retire puppet-qdr: Remove Project Content","message":"Retire puppet-qdr: Remove Project Content\n\nDepends-on: https://review.opendev.org/c/openstack/project-config/+/907951\nChange-Id: I77f7bc94b8d1b3459c64f44f5d18d5f8c747ae4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/89a7097f466486fdbafad00e00f00617568af17b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/89a7097f466486fdbafad00e00f00617568af17b"}]},"branch":"refs/heads/master"},"31aeb84bae9f82f8550000db1029b9d1c856d051":{"kind":"REWORK","_number":2,"created":"2024-02-22 01:46:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/907952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-qdr","ref":"refs/changes/52/907952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/52/907952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/52/907952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-qdr refs/changes/52/907952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-qdr refs/changes/52/907952/2"}}},"commit":{"parents":[{"commit":"e349e5febab128bf67794e07b91697a64e7a61f3","subject":"reno: Update master for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/e349e5febab128bf67794e07b91697a64e7a61f3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-06 07:04:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-22 01:46:27.000000000","tz":540},"subject":"Retire puppet-qdr: Remove Project Content","message":"Retire puppet-qdr: Remove Project Content\n\nDepends-on: https://review.opendev.org/c/openstack/project-config/+/907951\nChange-Id: I77f7bc94b8d1b3459c64f44f5d18d5f8c747ae4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/31aeb84bae9f82f8550000db1029b9d1c856d051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-qdr/commit/31aeb84bae9f82f8550000db1029b9d1c856d051"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
