)]}'
{"id":"openstack%2Fpuppet-rally~656130","triplet_id":"openstack%2Fpuppet-rally~master~I5d0f2f982daa0cb2d154620bfe122818eb8b72b4","project":"openstack/puppet-rally","branch":"master","topic":"bug/1826692","hashtags":[],"change_id":"I5d0f2f982daa0cb2d154620bfe122818eb8b72b4","subject":"Deprecate idle_timeout option","status":"MERGED","created":"2019-04-28 07:01:01.000000000","updated":"2019-04-29 06:41:10.000000000","submitted":"2019-04-29 06:41:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656130-1556520070713-cdd0ea5d","meta_rev_id":"fab5360cf247890767985397a4d085ec39a9a4b2","_number":656130,"virtual_id_number":656130,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-04-29 06:41:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-04-28 08:43:13.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-04-29 06:05:40.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-28 08:43:13.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-04-29 06:41:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6a1c6f035ecfa5547da063af6b20332a0ac7109","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 07:01:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b92311ee0a43ffbbf32c8cd40ee5d60f9d9797f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 07:19:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/30/656130/1/check/puppet-openstack-lint-ubuntu-bionic/d63a0ca/ : SUCCESS in 4m 10s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/30/656130/1/check/puppet-openstack-syntax-4-ubuntu-bionic/c14ae52/ : SUCCESS in 4m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/30/656130/1/check/puppet-openstack-syntax-5-ubuntu-bionic/6b936cc/ : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/30/656130/1/check/puppet-openstack-syntax-6-ubuntu-bionic/d4d4dcd/ : SUCCESS in 4m 56s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/30/656130/1/check/puppet-openstack-unit-4.8-centos-7/008f321/ : SUCCESS in 11m 30s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/30/656130/1/check/puppet-openstack-unit-5.5-centos-7/3909dde/ : SUCCESS in 10m 58s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/30/656130/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/2e4cd8a/ : SUCCESS in 8m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/30/656130/1/check/puppet-openstack-unit-latest-ubuntu-bionic/a70cee8/ : SUCCESS in 8m 56s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/30/656130/1/check/puppet-openstack-beaker-centos-7/593e76c/ : SUCCESS in 9m 49s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/30/656130/1/check/puppet-openstack-beaker-ubuntu-bionic/1ec97b9/ : SUCCESS in 7m 42s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/30/656130/1/check/build-openstack-releasenotes/47c7cb9/html/ : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"92481ae95a27624495ed52d0977f7317f87dfe47","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-04-28 08:43:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e0cb6bddec292279bf1d753a37242f9c0655e067","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-29 06:05:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"25498afad5dc82e6abca9c2309a79657682cd7a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 06:05:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1dd69937aea8cd461ba15d8bfb127e67d14225ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 06:41:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d5fa46f375049797ed0559b686315613b7c88787","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 06:41:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/30/656130/1/gate/puppet-openstack-lint-ubuntu-bionic/c41c1e4/ : SUCCESS in 3m 49s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/30/656130/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/6dcc2fd/ : SUCCESS in 5m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/30/656130/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/286497f/ : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/30/656130/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/2dd4aaf/ : SUCCESS in 3m 55s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/30/656130/1/gate/puppet-openstack-unit-4.8-centos-7/ec14475/ : SUCCESS in 12m 58s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/30/656130/1/gate/puppet-openstack-unit-5.5-centos-7/05c6de7/ : SUCCESS in 12m 21s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/30/656130/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/5773d7d/ : SUCCESS in 8m 10s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/30/656130/1/gate/puppet-openstack-beaker-centos-7/9e8acc5/ : SUCCESS in 6m 55s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/30/656130/1/gate/puppet-openstack-beaker-ubuntu-bionic/7c3eb5b/ : SUCCESS in 5m 54s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/30/656130/1/gate/build-openstack-releasenotes/8b2990a/html/ : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e690501b372134385ad30b4955cdf9d4e67c362","revisions":{"0e690501b372134385ad30b4955cdf9d4e67c362":{"kind":"REWORK","_number":1,"created":"2019-04-28 07:01:01.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/30/656130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/30/656130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/30/656130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/30/656130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/30/656130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/30/656130/1"}}},"commit":{"parents":[{"commit":"3a077381b0e8c6a498b25bdd0647124592d621b1","subject":"Move to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/3a077381b0e8c6a498b25bdd0647124592d621b1"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:00:12.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:00:12.000000000","tz":480},"subject":"Deprecate idle_timeout option","message":"Deprecate idle_timeout option\n\nThe idle_timeout parameter is deprecated, use connection_recycle_time\ninstead[1].\n\n[1]https://review.opendev.org/#/c/334182/\n\nChange-Id: I5d0f2f982daa0cb2d154620bfe122818eb8b72b4\nDepends-On: https://review.opendev.org/656106/\nCloses-Bug: #1826692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/0e690501b372134385ad30b4955cdf9d4e67c362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/0e690501b372134385ad30b4955cdf9d4e67c362"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
