)]}'
{"id":"openstack%2Fpuppet-rally~684888","triplet_id":"openstack%2Fpuppet-rally~master~Ifa4d52357696328fa5e1c6f6f37263fa9142c7b4","project":"openstack/puppet-rally","branch":"master","topic":"constraints","hashtags":[],"change_id":"Ifa4d52357696328fa5e1c6f6f37263fa9142c7b4","subject":"Update the constraints url","status":"MERGED","created":"2019-09-26 02:49:32.000000000","updated":"2019-09-27 03:19:08.000000000","submitted":"2019-09-27 03:17:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684888-1569554271652-e985b01a","meta_rev_id":"fe49085a365185170a8c0f3cf458596d3a0fd9c6","_number":684888,"virtual_id_number":684888,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 03:17:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-27 01:49:45.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-27 01:49:45.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 01:49:45.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-27 03:17:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"873c915613d5d0fe68b55d31ab0291e7a23860f8","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-26 02:49:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fe2bfa8fdac27aa72e956d22f33720f16180ee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 03:05:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7631e7ff521e461c8d1bba11698b03e8 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c79d8d752c347578edb593442025bb9 : SUCCESS in 5m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99cd07f4bf4f479fa5305aa7385ee405 : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/33f56ce8c8dd4abd8bdbabbab5b029a1 : SUCCESS in 12m 32s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d31cf4ab5f6a4d809f6be35fd3b79ca8 : SUCCESS in 10m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/366e169380294ad1b27ba1c42a11d49f : SUCCESS in 10m 44s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6088f7f6db1a47c18d0c5cb4dc3d2d71 : SUCCESS in 11m 22s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c89232ae14884cf0b24beeb39f0b3c33 : SUCCESS in 4m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c11ba24c749b445cb68941d64fc26ded : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ed893b79e19b5c536e98f7ca32f486078f39c0a5","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-27 01:49:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62b2b36e723ed239d87530cd2d090414c058f9fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:49:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a4d644719585d9afcf712481498732a19045a0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 03:17:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4dee01763c644a13ab2e70bbbee911a0 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc1062ff5af24795b74eb6aa39177476 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ff525a1075b40e4904a03786853ebf1 : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bab487e801334b83b018752321602004 : SUCCESS in 16m 24s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43e1d4a6eb9d41caa47b4df895a8bbc2 : SUCCESS in 11m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/608db3a4a56b4080bbf023203d202ba5 : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4103fac5e254ee9bf4c7fb7f770ae58 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"a70399010afec6e96d5c9a7277f7f8367be87f07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 03:17:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff165a25788e17a30e2e4b8aa3ae71189b462382","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 03:19:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/894d9cbc498f4c1987c8f7bc362e5b49 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b348f9d879cba3053a391ed26b37e99f7e167070","revisions":{"b348f9d879cba3053a391ed26b37e99f7e167070":{"kind":"REWORK","_number":1,"created":"2019-09-26 02:49:32.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/88/684888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/88/684888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/88/684888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/88/684888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/88/684888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/88/684888/1"}}},"commit":{"parents":[{"commit":"3072885d9bfcfe13c820bbeaf467b7595311bd71","subject":"Prepare Train M4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/3072885d9bfcfe13c820bbeaf467b7595311bd71"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 02:49:47.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-26 02:50:16.000000000","tz":480},"subject":"Update the constraints url","message":"Update the constraints url\n\nFor more detail, see http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: Ifa4d52357696328fa5e1c6f6f37263fa9142c7b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/b348f9d879cba3053a391ed26b37e99f7e167070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/b348f9d879cba3053a391ed26b37e99f7e167070"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
