)]}'
{"id":"openstack%2Fpuppet-rally~687047","triplet_id":"openstack%2Fpuppet-rally~stable%2Ftrain~Ie50a9e5dce370e01a7ec96a97f7b1b8d41e4f4b1","project":"openstack/puppet-rally","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"Ie50a9e5dce370e01a7ec96a97f7b1b8d41e4f4b1","subject":"Update .gitreview for stable/train","status":"MERGED","created":"2019-10-07 13:35:22.000000000","updated":"2019-10-10 20:08:04.000000000","submitted":"2019-10-10 20:08:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687047-1570738084200-9bbf562d","meta_rev_id":"a09ece2ce7aba925a11a9afc564ea43291f62551","_number":687047,"virtual_id_number":687047,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 20:08:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-10-10 19:19:52.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-10-07 21:11:32.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-10-10 19:19:52.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 21:11:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-10 19:19:52.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-10 20:08:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45f39229b28e74ba9c5e66d5ffc8a27451818003","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-10-07 13:35:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"912f0826ee53d58bc1177b295adbb414df1eed37","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-07 21:11:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4bab6163ba12f188ad5b19eaf69f5e5e5e71371b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-09 19:39:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9edb460760b379625c35d4acf3559a5eb06da89a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 20:03:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abbad17a495a43cbbf931dadb60766db : SUCCESS in 4m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/747a88b65f1d4330996639a10581f90f : SUCCESS in 4m 16s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9467ee2f97814db085f6b87e2e1d4071 : SUCCESS in 17m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e3dce440092445d3a26579e6bf4ffcad : FAILURE in 6m 44s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34079ce2096a4a5c86532aae88071450 : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bd5f4a2edca92dfba8f7bbb1b9a49c1578f54e2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-10 19:19:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d940278e5947da95a924d072f2115cf21792bfbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 19:45:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11c687ca036d4e5ca2218253d4b5bf6f : SUCCESS in 5m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d60f5efcf62b44f397ce0c38142e51ec : SUCCESS in 5m 23s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/49a1695db0414bf3bf74c3f38e2137de : SUCCESS in 17m 39s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a83a43632e5c419e89ad3dd5ef08931c : SUCCESS in 8m 07s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3d6b976bbc14d3b9153be291ae2fb91 : SUCCESS in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a460c65d8c7878f806d2561f464044a43e61ae7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 19:45:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b930b1ace2d79ff38d94836f1e165e307f4fa1c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:08:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3b9856f047a4542b7288d558cf3ad8c : SUCCESS in 3m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e107b0c90ee471399dec809500abc0a : SUCCESS in 4m 19s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/efdfb2bd4d824a94ba77e323617224de : SUCCESS in 17m 21s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8b1206d5c8534cb48a9d9ce4392ff1d7 : SUCCESS in 9m 48s","accounts_in_message":[],"_revision_number":1},{"id":"90ec5a469cd788c1885d7408e60b10e388bf7e3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:08:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c81ba0c63f60eee901091b338380ee965d2be862","revisions":{"c81ba0c63f60eee901091b338380ee965d2be862":{"kind":"REWORK","_number":1,"created":"2019-10-07 13:35:22.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/47/687047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/47/687047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/47/687047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/47/687047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/47/687047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/47/687047/1"}}},"commit":{"parents":[{"commit":"3e149ef166d03add395e62c1fe33ff94a4e6313a","subject":"Merge \"Update the constraints url\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/3e149ef166d03add395e62c1fe33ff94a4e6313a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-07 13:35:20.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-07 13:35:20.000000000","tz":0},"subject":"Update .gitreview for stable/train","message":"Update .gitreview for stable/train\n\nChange-Id: Ie50a9e5dce370e01a7ec96a97f7b1b8d41e4f4b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/c81ba0c63f60eee901091b338380ee965d2be862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/c81ba0c63f60eee901091b338380ee965d2be862"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
