)]}'
{"id":"openstack%2Fpuppet-rally~698154","triplet_id":"openstack%2Fpuppet-rally~master~I6412331575b4878f4767b3656890d7127c546674","project":"openstack/puppet-rally","branch":"master","topic":"prep-release-ussuri-1","hashtags":[],"change_id":"I6412331575b4878f4767b3656890d7127c546674","subject":"Prepare Ussuri M1","status":"MERGED","created":"2019-12-10 06:25:50.000000000","updated":"2019-12-10 16:49:10.000000000","submitted":"2019-12-10 16:49:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698154-1575996550467-5614962f","meta_rev_id":"28d742d17475f76892d986153e8424ed35e56ffc","_number":698154,"virtual_id_number":698154,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 16:49:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-12-10 16:37:10.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-12-10 16:37:10.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 16:37:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-10 16:49:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7b842165bdddc4f1e7245b276c43d7e39dce0ae","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-12-10 06:25:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7964d9505ac71bff343917143a154934ffe7963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 06:36:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/0d2c32d32848484abe6dff8ba7c3e8c2 : SUCCESS in 2m 12s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c9b55fa628f449096d7941bc7b6d67f : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"750ffd71d07fecd65ccfb81a1e253f5b08f0e833","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-10 16:37:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5434a5a8bad39c3925631072b2d1b0218ccf2e6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:37:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c10d7664cbb60c483639acdbc7badebb181f5c77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:49:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"018be3e99d316f4b97df7c19a3c2692882757dca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:49:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/7c9f2acdd44e4f32ac4df84ce233f373 : SUCCESS in 2m 37s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87ff5e7583144723bc495f66a3c83e9c : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ce023ecd62ea099c68d3f7a661a03693a186258f","revisions":{"ce023ecd62ea099c68d3f7a661a03693a186258f":{"kind":"REWORK","_number":1,"created":"2019-12-10 06:25:50.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/54/698154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/54/698154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/54/698154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/54/698154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/54/698154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/54/698154/1"}}},"commit":{"parents":[{"commit":"ba8c937d3cfc0bf325a1986cee2d1303f031159d","subject":"Convert all class usage to relative names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/ba8c937d3cfc0bf325a1986cee2d1303f031159d"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-12-10 06:24:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-12-10 06:24:49.000000000","tz":480},"subject":"Prepare Ussuri M1","message":"Prepare Ussuri M1\n\nUpdate the version metadata for Ussuri milestone M1\n\nChange-Id: I6412331575b4878f4767b3656890d7127c546674\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/ce023ecd62ea099c68d3f7a661a03693a186258f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/ce023ecd62ea099c68d3f7a661a03693a186258f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
