)]}'
{"id":"openstack%2Fpuppet-rally~715822","triplet_id":"openstack%2Fpuppet-rally~master~I6a4e277cddf1f82a21e8f73f16b14c819e305806","project":"openstack/puppet-rally","branch":"master","hashtags":[],"change_id":"I6a4e277cddf1f82a21e8f73f16b14c819e305806","subject":"Add CentOS 8 and Puppet 6 in metadata","status":"MERGED","created":"2020-03-30 07:35:37.000000000","updated":"2020-03-31 15:53:23.000000000","submitted":"2020-03-31 15:53:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715822-1585670003589-6894be15","meta_rev_id":"6c74e0462e277bd62f253e63c3abd9e906f9edea","_number":715822,"virtual_id_number":715822,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 15:53:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-30 13:54:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-31 15:41:51.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-31 15:41:51.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 13:54:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-31 15:41:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-31 15:53:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"308dfb87c49acb11cd8d266f895efaaf98898d9a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-30 07:35:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"745c348b128b523b8a361ab03e4d53db5196b375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 07:44:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/9ad15e73e18f493b9085d90c604f0c80 : SUCCESS in 3m 01s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f8d3522d9164efcbe66d50fa9031de8 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"1d337e4e37004fd9b7372e1516e5e18852ec7d9f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-03-30 07:52:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a89807e64e87903313bd4951349261a972794486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 08:01:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/81c5abe304e8429682186c57c143749f : SUCCESS in 3m 20s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95eb393b2ac942f2a233efbc7252eba1 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"6d9193b6bda959285c65dcfab317d8015bb34d5c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-30 13:54:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"03d6604bd1a663d5cf5ca078ce7e65376c26d475","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-31 15:41:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"69b1e4572048e84c6ac42891f965ec9ee05a81d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:42:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"57f11402b189c17f2dc373d0466425b9bacd69cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:53:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/a47856342f864e76a8310b94b951035d : SUCCESS in 4m 08s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80a124c375244c60b2cdf2893ce08168 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"981cdf731ab3b0d6fe52662d8896618ee46ce4d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:53:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"06f7bd64a86e911ced4541a0808b64bc5fdf18cc","revisions":{"a15f35b322bd540b0b0255a29bcdd7a012fef7ee":{"kind":"REWORK","_number":1,"created":"2020-03-30 07:35:37.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/22/715822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/22/715822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/22/715822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/22/715822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/22/715822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/22/715822/1"}}},"commit":{"parents":[{"commit":"bddd08d31952e351bcf551e0b1db7e067bbce5f5","subject":"Deprecate min_pool_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/bddd08d31952e351bcf551e0b1db7e067bbce5f5"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:35:30.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:35:30.000000000","tz":120},"subject":"Add CentOS 8 and Puppet 6 in metadata","message":"Add CentOS 8 and Puppet 6 in metadata\n\nChange-Id: I6a4e277cddf1f82a21e8f73f16b14c819e305806\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/a15f35b322bd540b0b0255a29bcdd7a012fef7ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/a15f35b322bd540b0b0255a29bcdd7a012fef7ee"}]},"branch":"refs/heads/master"},"06f7bd64a86e911ced4541a0808b64bc5fdf18cc":{"kind":"REWORK","_number":2,"created":"2020-03-30 07:52:12.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/22/715822/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/22/715822/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/22/715822/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/22/715822/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/22/715822/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/22/715822/2"}}},"commit":{"parents":[{"commit":"bddd08d31952e351bcf551e0b1db7e067bbce5f5","subject":"Deprecate min_pool_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/bddd08d31952e351bcf551e0b1db7e067bbce5f5"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:35:30.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-03-30 07:52:10.000000000","tz":0},"subject":"Add CentOS 8 and Puppet 6 in metadata","message":"Add CentOS 8 and Puppet 6 in metadata\n\nChange-Id: I6a4e277cddf1f82a21e8f73f16b14c819e305806\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/06f7bd64a86e911ced4541a0808b64bc5fdf18cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/06f7bd64a86e911ced4541a0808b64bc5fdf18cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
