)]}'
{"id":"openstack%2Fpuppet-rally~726685","triplet_id":"openstack%2Fpuppet-rally~master~I29b67b26ab1f013da36ef3b5004574461e456b1e","project":"openstack/puppet-rally","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"I29b67b26ab1f013da36ef3b5004574461e456b1e","subject":"[ussuri][goal] Add CONTRIBUTING.rst","status":"MERGED","created":"2020-05-10 20:51:42.000000000","updated":"2020-05-12 12:45:05.000000000","submitted":"2020-05-12 12:45:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726685-1589287505917-d1a6a7d1","meta_rev_id":"d8a36d512a22c286e0e0dcd319f3d5522c2ac412","_number":726685,"virtual_id_number":726685,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 12:45:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-05-12 12:33:15.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-11 22:21:17.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-05-12 12:33:15.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 22:21:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-12 12:33:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-12 12:45:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3f76b0ed91616f3486e93ccc0316079031f0050","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-10 20:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13d4ef5735c8123475b3141e81da9ffbbb138d8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 21:07:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ce6dd1d502f4d289c52003c75ec4f07 : SUCCESS in 4m 44s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9133ff014c8040da9b0ac57fcf241290 : SUCCESS in 13m 34s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/017af656a9b14c9c857d63d90f389960 : SUCCESS in 8m 47s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c345414868684c568c625d9a7ca6158c : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee7d8f0291e259d9a94f9161155a3b067b263bee","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-11 22:21:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5efe119e5003ce63e925494cfbb5d8b52ea5c2ef","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-12 12:33:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fbc0639b18de994b5e09f68ca26790455f62374b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 12:33:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e39a0f1c23ab7a4b7f9da60ad29360ab136a927","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 12:45:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"01a11e8f4d3aa2cf8fc595890901a56900e6f5df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 12:45:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/399387a875aa4c24831e21f45fa883eb : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d7d86d648f428d3e7ece914d105f293f4b04e74","revisions":{"7d7d86d648f428d3e7ece914d105f293f4b04e74":{"kind":"REWORK","_number":1,"created":"2020-05-10 20:51:42.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/85/726685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/85/726685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/85/726685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/85/726685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/85/726685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/85/726685/1"}}},"commit":{"parents":[{"commit":"cbddb909cb20c5f6a3f18fc323b91d8bda450a10","subject":"Bump openstackdocstheme to 1.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/cbddb909cb20c5f6a3f18fc323b91d8bda450a10"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-10 20:51:34.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-10 20:51:34.000000000","tz":120},"subject":"[ussuri][goal] Add CONTRIBUTING.rst","message":"[ussuri][goal] Add CONTRIBUTING.rst\n\nStory: 2007236\nTask: 38552\n\nDepends-On: https://review.opendev.org/#/c/726509/\nChange-Id: I29b67b26ab1f013da36ef3b5004574461e456b1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/7d7d86d648f428d3e7ece914d105f293f4b04e74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/7d7d86d648f428d3e7ece914d105f293f4b04e74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
