)]}'
{"id":"openstack%2Fpuppet-rally~748080","triplet_id":"openstack%2Fpuppet-rally~master~Icb359deceaadbb01a35f05e8fba2c8a1fd265019","project":"openstack/puppet-rally","branch":"master","topic":"bug/1892952","hashtags":[],"change_id":"Icb359deceaadbb01a35f05e8fba2c8a1fd265019","subject":"Add mysql_enable_ndb option","status":"MERGED","created":"2020-08-26 03:54:20.000000000","updated":"2020-08-26 21:07:38.000000000","submitted":"2020-08-26 21:05:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748080-1598475900578-98515791","meta_rev_id":"101b6c68452e4f6314a70fbe01bb1f49d5a0653f","_number":748080,"virtual_id_number":748080,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 21:05:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-26 20:43:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-26 15:47:28.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-26 20:43:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 15:47:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-26 20:43:11.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-26 21:05:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94b6cc18dff6cd6caa8cd36340c723044aabfb24","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-26 03:54:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f42b41ad6c48ced058167d28c85752f569e473a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-26 03:57:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"74f975ea6cd86815ff3c87dd89e52b965fa16456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 04:22:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4ead20614554f778588ad58f995b8ed : SUCCESS in 4m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c815dfa56d3849d08eeca63e01a7b3ae : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95b214d284314e809df9423515851f95 : SUCCESS in 4m 10s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6b4c3946ac334b67b6bd9d853589c1ac : SUCCESS in 9m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/43170e82ef8d470994b4cc2a25e265dc : SUCCESS in 11m 32s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/496db93c0887419cb02082ae67e3e408 : SUCCESS in 8m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1aac56295cb047b793f8cc3514b62e30 : SUCCESS in 9m 27s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/8fce9443b6794cadbf1117aecb3155cf : SUCCESS in 8m 46s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5726ccf053948f2bfba6b0162a6f4bd : SUCCESS in 6m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a74f986aa5854823abb87ffe1b335494 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":2},{"id":"f4a6db40ef3f3da1defe6bc0775aa7f4804e3bba","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-26 15:47:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fff0c227f5ceb6f9006fd2902696ad360c88b00f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-26 20:43:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"47536db140ae7171a578f92cc8b0db4266e368a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:43:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d56e21cde883a92635c008ff0b8e9685e3ceae1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 21:05:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d26bca9809384ed9afb5a739acbbfbde : SUCCESS in 3m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bac13597126140a69c24d25d5287f711 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4822be3071fc413caf0680c3ed083504 : SUCCESS in 3m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/dff9bdd245ea4cb99cf9281f47dfe356 : SUCCESS in 13m 46s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d86d3cd9f2094a04a075904da9861bf2 : SUCCESS in 11m 16s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cf333215c804082989325147827ba21 : SUCCESS in 8m 12s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/813eb7db03a5480f82636fd646bdb494 : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d504ab4697e4db29e243d9bc547aad1 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6f135626a9bcb306f7a9c6c2d5c172300d624324","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 21:05:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6a50ae36559a0cfd8bdf61e0a6f287ae53740770","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 21:07:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fa117a390244ca09b3a44e6b59971cc : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b796c0f67b5159dc9e473fd60abedcc7292aa31","revisions":{"663e45bb641e8c5b44cc74f6bfa0bce3f196f822":{"kind":"REWORK","_number":1,"created":"2020-08-26 03:54:20.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/80/748080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/80/748080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/80/748080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/80/748080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/80/748080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/80/748080/1"}}},"commit":{"parents":[{"commit":"872a19e0831e73603d0db3b3c7127a32ed5692f8","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/872a19e0831e73603d0db3b3c7127a32ed5692f8"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"subject":"Add mysql_enable_ndb option","message":"Add mysql_enable_ndb option\n\nAdd mysql_enable_ndb parameter to select mysql storage engine.\n\nChange-Id: Icb359deceaadbb01a35f05e8fba2c8a1fd265019\nCloses-Bug: #1892952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/663e45bb641e8c5b44cc74f6bfa0bce3f196f822"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/663e45bb641e8c5b44cc74f6bfa0bce3f196f822"}]},"branch":"refs/heads/master"},"5b796c0f67b5159dc9e473fd60abedcc7292aa31":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-26 03:57:22.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/80/748080/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/80/748080/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/80/748080/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/80/748080/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/80/748080/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/80/748080/2"}}},"commit":{"parents":[{"commit":"872a19e0831e73603d0db3b3c7127a32ed5692f8","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/872a19e0831e73603d0db3b3c7127a32ed5692f8"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:53:17.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-26 03:56:27.000000000","tz":480},"subject":"Add mysql_enable_ndb option","message":"Add mysql_enable_ndb option\n\nAdd mysql_enable_ndb parameter to select mysql storage engine.\n\nChange-Id: Icb359deceaadbb01a35f05e8fba2c8a1fd265019\nDepends-On: https://review.opendev.org/#/c/748067\nCloses-Bug: #1892952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/5b796c0f67b5159dc9e473fd60abedcc7292aa31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/5b796c0f67b5159dc9e473fd60abedcc7292aa31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
