)]}'
{"id":"openstack%2Fpuppet-rally~749133","triplet_id":"openstack%2Fpuppet-rally~master~Ic316338cd6f0127f24d14b05c4fcb75fa57060c4","project":"openstack/puppet-rally","branch":"master","hashtags":[],"change_id":"Ic316338cd6f0127f24d14b05c4fcb75fa57060c4","subject":"Change beaker to litmus","status":"MERGED","created":"2020-08-31 19:54:36.000000000","updated":"2020-09-08 18:54:29.000000000","submitted":"2020-09-08 18:54:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749133-1599591269158-9a8fb52f","meta_rev_id":"4de81c5e4e97f0388ea1280b27fd3f48117430e3","_number":749133,"virtual_id_number":749133,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 18:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-08 18:28:42.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-08 18:26:24.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-08 18:28:42.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 18:26:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-09-08 18:28:42.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-08 18:54:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4de9e3974acfd69859b0a2e671b92105f5debad4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 19:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e3b198def3643d870acbdc5948c0fd724f95ee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 20:10:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07d69d026dc94225b4bcb8a832b78b6a : SUCCESS in 2m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f630747b2faf4d30a18908cdbbfe4b1b : SUCCESS in 2m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/123b65ff8c324862b63e6ed4bbc94071 : SUCCESS in 2m 32s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4fbbdf9edeec478095dbcf7dfbc99adb : SUCCESS in 8m 09s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/6508cdda14164be199fd700271ffec57 : SUCCESS in 8m 58s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/090b20e45d7741d0ad947d63713d8016 : SUCCESS in 6m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d74670317f0141358f69a03f5114bf0f : SUCCESS in 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92feb87a8b8e62da4116c075dc88b0ba3137a49e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-08 18:26:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2102e476c9012cf64124c5d874d626f2eb53f5c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 18:28:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"07030606144a099701071eb57ca4ab699d5a0a6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 18:31:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad9a590c50c51f8354db0b9638b6dd521075d944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 18:54:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9faa95329b249be9bbefb643865cd6c : SUCCESS in 3m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61212c22b52d49cf8ea33d4124c50e37 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb2dafe644fa4e87b7dfc5f61ea34c1e : SUCCESS in 3m 30s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7525793e53f24aae89c62fe7df187a6d : SUCCESS in 13m 40s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5dea1705dbd9455c9200588664f19538 : SUCCESS in 10m 16s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a8cdb5d181a4d748ac6b0c9bd44562b : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4de81c5e4e97f0388ea1280b27fd3f48117430e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 18:54:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9eeaf5cade268a82006ab9d4369c6533e0103d05","revisions":{"9eeaf5cade268a82006ab9d4369c6533e0103d05":{"kind":"REWORK","_number":1,"created":"2020-08-31 19:54:36.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/33/749133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/33/749133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/33/749133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/33/749133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/33/749133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/33/749133/1"}}},"commit":{"parents":[{"commit":"9864f0782501eda492396b14469940559500feca","subject":"Merge \"Do not validate database_connection format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/9864f0782501eda492396b14469940559500feca"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:54:21.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-08-31 19:54:21.000000000","tz":120},"subject":"Change beaker to litmus","message":"Change beaker to litmus\n\nIt\u0027s not executed, but let\u0027s cleanup.\n\nChange-Id: Ic316338cd6f0127f24d14b05c4fcb75fa57060c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/9eeaf5cade268a82006ab9d4369c6533e0103d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/9eeaf5cade268a82006ab9d4369c6533e0103d05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
