)]}'
{"id":"openstack%2Fpuppet-rally~750261","triplet_id":"openstack%2Fpuppet-rally~master~Ife74a2070b4fcdcfe95c194ee158a315c7e5621f","project":"openstack/puppet-rally","branch":"master","topic":"prep-release-victoria-m3","hashtags":[],"change_id":"Ife74a2070b4fcdcfe95c194ee158a315c7e5621f","subject":"Prepare Victoria M3","status":"MERGED","created":"2020-09-08 01:18:47.000000000","updated":"2020-09-08 13:11:40.000000000","submitted":"2020-09-08 13:11:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750261-1599570700533-51e315b5","meta_rev_id":"f9ca7dd0b2d730a24d532d2bfdd1f92709139042","_number":750261,"virtual_id_number":750261,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-08 13:11:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-08 13:01:17.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-08 13:01:17.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 13:01:17.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-08 13:11:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16ef7638d4a3087c6416755c775d8a1817023c10","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-09-08 01:18:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8454e8cd455d94de1de968a011b3454f7daf288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 01:26:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/09934b535a674de1841dd96c09af179b : SUCCESS in 2m 04s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6369a09e4ed4426d9c7997ba7565a7e0 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a45e3f280c0beb62014800b7bed705096a548d4a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-08 13:01:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b91f8b1e2e9bec600cd95a9dbee96e63cf95817d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:01:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92fb68c6e6bb8e0efa7c34c4e8500d76f59c2935","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:11:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f56397ada8633bd1f23ce4b8d7d6e42fdd0764d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 13:11:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/527fda1315ad4b50a09d26b13fb642af : SUCCESS in 3m 27s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a124e39cd2624b9ea1f3fe4c6e015bdc : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f08d660a269d065b0313eb56816fb8366f80c907","revisions":{"f08d660a269d065b0313eb56816fb8366f80c907":{"kind":"REWORK","_number":1,"created":"2020-09-08 01:18:47.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/61/750261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/61/750261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/61/750261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/61/750261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/61/750261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/61/750261/1"}}},"commit":{"parents":[{"commit":"9864f0782501eda492396b14469940559500feca","subject":"Merge \"Do not validate database_connection format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/9864f0782501eda492396b14469940559500feca"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-09-08 01:18:03.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-09-08 01:18:03.000000000","tz":480},"subject":"Prepare Victoria M3","message":"Prepare Victoria M3\n\nUpdate the version metadata for Victoria milestone M3\n\nChange-Id: Ife74a2070b4fcdcfe95c194ee158a315c7e5621f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/f08d660a269d065b0313eb56816fb8366f80c907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/f08d660a269d065b0313eb56816fb8366f80c907"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
