)]}'
{"id":"openstack%2Fpuppet-rally~756944","triplet_id":"openstack%2Fpuppet-rally~master~I3b58fc0c3bcc28e32467307a87800f1e2e16d67a","project":"openstack/puppet-rally","branch":"master","topic":"missing-deps","hashtags":[],"change_id":"I3b58fc0c3bcc28e32467307a87800f1e2e16d67a","subject":"Include deps class in unit test for mysql","status":"MERGED","created":"2020-10-09 02:00:41.000000000","updated":"2020-10-09 03:06:32.000000000","submitted":"2020-10-09 03:06:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756944-1602212792034-25194bd4","meta_rev_id":"199bc9de691c9151d281e94a40632371c2545e2f","_number":756944,"virtual_id_number":756944,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-09 03:06:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-09 02:46:50.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-10-09 02:46:50.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 03:06:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5f734ea4ace02625370c941076c59c50ec6452e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-09 02:00:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9c77d1a67482a3c17f395457e4e0d6d80354a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:21:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5fd6642af7c42d99ede0f7de4ca1c8c : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5073fff88ab7489eb63763f65b66b993 : SUCCESS in 2m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/a480e767d18247a4bee6a21ec7fc3c18 : SUCCESS in 9m 11s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/370e88fdbf654681874ce6209f4df97e : SUCCESS in 7m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2714b32e2fd54871baa84759dab5fc61 : SUCCESS in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"881dd05cbbf236a632480fbaeb21597a78450405","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-09 02:46:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"465f8824ad92d20d3acda7fc486717f5500d210a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:47:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9f11a53113f67ffcc4e19d0c0ec8b6770c4dda2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:06:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3f3f8d45e9041c4a40b190e5ee33943 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/069eb4bd70dd4ac8bc552b97a167c592 : SUCCESS in 2m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/53f19ed314e348a98086708cc419934c : SUCCESS in 13m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eee139bfb98144378d1d43948cfbeb12 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":1},{"id":"aa7d0402e713723ccc5251d0c5e013020ce4f214","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:06:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a9e7c8e4511039442e874027da7bcafdc7a3b75","revisions":{"6a9e7c8e4511039442e874027da7bcafdc7a3b75":{"kind":"REWORK","_number":1,"created":"2020-10-09 02:00:41.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/44/756944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/44/756944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/44/756944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/44/756944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/44/756944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/44/756944/1"}}},"commit":{"parents":[{"commit":"f1965b847173fea47039ce679833ab88e1b53060","subject":"Prepare Victoria RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/f1965b847173fea47039ce679833ab88e1b53060"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 01:59:55.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 01:59:55.000000000","tz":480},"subject":"Include deps class in unit test for mysql","message":"Include deps class in unit test for mysql\n\nChange-Id: I3b58fc0c3bcc28e32467307a87800f1e2e16d67a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/6a9e7c8e4511039442e874027da7bcafdc7a3b75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/6a9e7c8e4511039442e874027da7bcafdc7a3b75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
