)]}'
{"id":"openstack%2Fpuppet-rally~819969","triplet_id":"openstack%2Fpuppet-rally~master~I3831ad848f1d7026fbcd6c19269ae5deecb4b376","project":"openstack/puppet-rally","branch":"master","topic":"prep-release-yoga-m2","attention_set":{},"removed_from_attention_set":{"9414":{"account":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"last_update":"2022-01-07 14:37:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3831ad848f1d7026fbcd6c19269ae5deecb4b376","subject":"Prepare Yoga M2","status":"MERGED","created":"2021-12-01 06:43:38.000000000","updated":"2022-01-07 14:37:32.000000000","submitted":"2022-01-07 14:37:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819969-prep-release-yoga-m2","meta_rev_id":"89526e7c81d9fd13fdc5d157eefcd8a76861d3f0","_number":819969,"virtual_id_number":819969,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-07 14:37:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-01-07 14:22:49.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-01-07 14:37:32.000000000","post_submit":true,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-01-07 14:22:49.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-01 06:54:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-13 15:38:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-07 14:22:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"55450f3ff94e8acc4d56b2dc74aca1a8cab95c60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-12-01 06:43:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e428f7be515ea661689b373c67253f6728f0d0aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 06:54:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/16a97c5a419047d29c18494f892bc676 : SUCCESS in 2m 35s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00ea458ce1fb472b803b7b5cc14dc404 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"90dcb642ce3515b2e01387f0ed8d8d90a5a85ce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-12-03 08:56:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a5169e99d496b6186f2147879cff3eefbc9abeec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 09:05:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/f90dc24019f7434fa70a38872ac626f6 : SUCCESS in 2m 29s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1b07b3689354780974cbbe2b28ceb33 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":2},{"id":"16669e0338a7bc66738a0a2c26f2ad6b71c05f5e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-13 15:38:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c0aa163b2260a4653336a06ed8e4f25dc11a2b4f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-03 02:42:37.000000000","message":"Topic prep-release-yoga-m1 removed","accounts_in_message":[],"_revision_number":2},{"id":"0b76a8674f25ef4c8c49f4c140b59f6174d57d2b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-03 02:42:40.000000000","message":"Topic set to prep-release-yoga-m2","accounts_in_message":[],"_revision_number":2},{"id":"114595306e3711d6b3e72025180e473990401689","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-07 14:22:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"20f4b3c0ff7edd5876218780c76c7ceb871a1455","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-07 14:22:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"18e44364aa98344d7689f127a1b2c7067dc24078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:31:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/c8214e45361b4faaabfc44ded42c49e5 : SUCCESS in 2m 20s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e6bcae0890046c48c1bea9874d26f6d : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":3},{"id":"0f23a619560131c26ab69873c0247318d616786b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:32:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6de89421c39f9cfde36758ff8003e4784345123d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:37:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/fa7de50ed3324b9ca2571bfcfed984c9 : SUCCESS in 3m 05s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae4d35f1c70a4cee822102786c141b11 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":3},{"id":"8b7e739244f39a4a867f10b8b3d0b0cd70a88dd8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 14:37:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"26cfdba4b042d48e0de0ab8e48b3ea2a3609d058","revisions":{"f1203e92997b6b7bef35d8459cc56f701e394a00":{"kind":"REWORK","_number":1,"created":"2021-12-01 06:43:38.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/69/819969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/69/819969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/1"}}},"commit":{"parents":[{"commit":"7c7bb865b7e0e4d8e06a660fc657bbd668ca846b","subject":"Bump puppetlabs-inifile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/7c7bb865b7e0e4d8e06a660fc657bbd668ca846b"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:53.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:53.000000000","tz":480},"subject":"Prepare Yoga M1","message":"Prepare Yoga M1\n\nUpdate the version metadata for Yoga milestone M1\n\nChange-Id: I3831ad848f1d7026fbcd6c19269ae5deecb4b376\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/f1203e92997b6b7bef35d8459cc56f701e394a00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/f1203e92997b6b7bef35d8459cc56f701e394a00"}]},"branch":"refs/heads/master"},"a3815e7be67d1be24aecc304805f3854029e6cf5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-03 08:56:21.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/69/819969/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/69/819969/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/2"}}},"commit":{"parents":[{"commit":"7c7bb865b7e0e4d8e06a660fc657bbd668ca846b","subject":"Bump puppetlabs-inifile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/7c7bb865b7e0e4d8e06a660fc657bbd668ca846b"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:53.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-02 06:00:39.000000000","tz":480},"subject":"Prepare Yoga M2","message":"Prepare Yoga M2\n\nUpdate the version metadata for Yoga milestone M2\n\nChange-Id: I3831ad848f1d7026fbcd6c19269ae5deecb4b376\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/a3815e7be67d1be24aecc304805f3854029e6cf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/a3815e7be67d1be24aecc304805f3854029e6cf5"}]},"branch":"refs/heads/master"},"26cfdba4b042d48e0de0ab8e48b3ea2a3609d058":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-01-07 14:22:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/819969/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/69/819969/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/69/819969/3"}}},"commit":{"parents":[{"commit":"27307da3be837a60b650101b73c7cb0d262aaa0b","subject":"Load libraries in a single place","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/27307da3be837a60b650101b73c7cb0d262aaa0b"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-12-01 06:42:53.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-07 14:22:45.000000000","tz":0},"subject":"Prepare Yoga M2","message":"Prepare Yoga M2\n\nUpdate the version metadata for Yoga milestone M2\n\nChange-Id: I3831ad848f1d7026fbcd6c19269ae5deecb4b376\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/26cfdba4b042d48e0de0ab8e48b3ea2a3609d058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/26cfdba4b042d48e0de0ab8e48b3ea2a3609d058"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
