)]}'
{"id":"openstack%2Fpuppet-rally~843053","triplet_id":"openstack%2Fpuppet-rally~master~I7cd90ba696a98d2d12f876680aa9e5b4390fb6e1","project":"openstack/puppet-rally","branch":"master","topic":"prep-release-zed-m1","attention_set":{},"removed_from_attention_set":{"9414":{"account":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"last_update":"2022-05-24 03:06:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7cd90ba696a98d2d12f876680aa9e5b4390fb6e1","subject":"Prepare Zed M1","status":"ABANDONED","created":"2022-05-24 02:01:05.000000000","updated":"2022-05-24 03:06:40.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"10875585074654e4bfe6496bf285c0777e0414e9","_number":843053,"virtual_id_number":843053,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-24 02:31:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2022-05-24 02:25:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-24 02:25:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-05-24 02:31:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6a1f5bc104930025c139b6f0c5cdd7d8aa773e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2022-05-24 02:01:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"863d99e07cf18de84ac0725e83bf06458fed3933","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-24 02:25:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39b1b725b9f2611c221801efdf54085c7cc188a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 02:31:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/89e2e87f95584aadabb0f0dd4f9729a2 : SUCCESS in 2m 29s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d282d5966d0422995b2c3fa3566c058 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"10875585074654e4bfe6496bf285c0777e0414e9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2022-05-24 03:06:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d485cad323a039f7a66123f3c39239ace75b0b16","revisions":{"d485cad323a039f7a66123f3c39239ace75b0b16":{"kind":"REWORK","_number":1,"created":"2022-05-24 02:01:05.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/53/843053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/53/843053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/53/843053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/53/843053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/53/843053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/53/843053/1"}}},"commit":{"parents":[{"commit":"859482b8930174ed900c439159cf9bfa660876d1","subject":"Update master for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/859482b8930174ed900c439159cf9bfa660876d1"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2022-05-24 02:00:24.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2022-05-24 02:00:24.000000000","tz":480},"subject":"Prepare Zed M1","message":"Prepare Zed M1\n\nUpdate the version metadata for Zed milestone M1\n\nChange-Id: I7cd90ba696a98d2d12f876680aa9e5b4390fb6e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/d485cad323a039f7a66123f3c39239ace75b0b16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/d485cad323a039f7a66123f3c39239ace75b0b16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
