)]}'
{"id":"openstack%2Fpuppet-rally~869628","triplet_id":"openstack%2Fpuppet-rally~master~I34570e7c4131c3f06a1da23e9f2e57eec75ff25a","project":"openstack/puppet-rally","branch":"master","topic":"bug/2002372","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-26 17:35:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I34570e7c4131c3f06a1da23e9f2e57eec75ff25a","subject":"Replace outdated command to create rally db","status":"MERGED","created":"2023-01-10 04:45:09.000000000","updated":"2023-02-26 17:35:04.000000000","submitted":"2023-02-26 17:35:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869628-bug/2002372","meta_rev_id":"626ad16ec1c5ef5c76dec57f3c4a15a625f14ed2","_number":869628,"virtual_id_number":869628,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-26 17:35:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-01-10 19:58:24.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-02-26 17:10:09.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-02-26 17:10:09.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 05:06:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-10 19:58:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-02-26 17:10:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7512c9bcac6d2aa45f8642fcc36a2e73413632db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-10 04:45:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dce6813777f50f2b6799651eb5f5062cf09d3631","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-10 04:45:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"30008a2dc96f7135e267a2332a2448d1d5d215ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-10 04:55:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9afe09b61fa57b5bf762c8398b9de5171cb5a9a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 05:06:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9af64b6fab1b4853902e4e16fe5cd8e8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee53388581ad4520acee91d47f1b4dd7 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/516f8ab038ff4bce99dfea894a582366 : SUCCESS in 2m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f28cba816dd0475592ac6fa522662ab3 : SUCCESS in 2m 30s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/8deb110028914af78a11df29c1471521 : SUCCESS in 8m 15s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5835d123b8594ce98e3f001e6f382cfa : SUCCESS in 5m 54s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b09e39c847114b699bf538b5622eb8cf : SUCCESS in 5m 58s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/483e495366d54258932482bf0fbe6d25 : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"497407b4bf822778f3634f0f6e65cc587de01209","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-01-10 19:58:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8a24326ad053618249af9ac28c00a558cb8ae091","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-02-26 17:10:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"833c1a4032df0edcd6e3b055d28f7e1a4a4c125c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 17:11:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"36b693397b914c8db33252cf84477b630ae96b27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 17:35:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be7176fbdb584e7da27f47777385a764\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5631e03dafb64b2eb6ebcc2a27765aba : SUCCESS in 2m 49s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35f2ada4634d47838248a04036ebd7c7 : SUCCESS in 2m 22s\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/16eaf0b1e64144f8999dbcb76baa270f : SUCCESS in 11m 08s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9d3839f4c26428dae1765bb14c3a5cd : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":3},{"id":"626ad16ec1c5ef5c76dec57f3c4a15a625f14ed2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-26 17:35:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"16a35f2fc6fdcaf039f7c483449b9b9230324659","revisions":{"0f3c65cdb6fcb406f722cb4c4fa4201927aea2d4":{"kind":"REWORK","_number":1,"created":"2023-01-10 04:45:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/869628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/28/869628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/1"}}},"commit":{"parents":[{"commit":"a179d50c7468ad7f9c871db499c872b4805c73ba","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/a179d50c7468ad7f9c871db499c872b4805c73ba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-10 04:44:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-10 04:44:02.000000000","tz":540},"subject":"Replace outdated command to create rally db","message":"Replace outdated command to create rally db\n\nThe rally-manage db command was already replaced by the rally db\ncommand[1]. This fixes the outdated command still used.\n\n[1] https://review.opendev.org/c/openstack/rally/+/681224\n\nChange-Id: I34570e7c4131c3f06a1da23e9f2e57eec75ff25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/0f3c65cdb6fcb406f722cb4c4fa4201927aea2d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/0f3c65cdb6fcb406f722cb4c4fa4201927aea2d4"}]},"branch":"refs/heads/master"},"eccc38c4f0d52d6dccda6f672feed9ee463d0bc8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-01-10 04:45:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/869628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/28/869628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/2"}}},"commit":{"parents":[{"commit":"a179d50c7468ad7f9c871db499c872b4805c73ba","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/a179d50c7468ad7f9c871db499c872b4805c73ba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-10 04:44:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-10 04:45:12.000000000","tz":540},"subject":"Replace outdated command to create rally db","message":"Replace outdated command to create rally db\n\nThe rally-manage db command was already replaced by the rally db\ncommand[1]. This fixes the outdated command still used.\n\n[1] https://review.opendev.org/c/openstack/rally/+/681224\n\nCloses-Bug: #2002372\nChange-Id: I34570e7c4131c3f06a1da23e9f2e57eec75ff25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/eccc38c4f0d52d6dccda6f672feed9ee463d0bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/eccc38c4f0d52d6dccda6f672feed9ee463d0bc8"}]},"branch":"refs/heads/master"},"16a35f2fc6fdcaf039f7c483449b9b9230324659":{"kind":"REWORK","_number":3,"created":"2023-01-10 04:55:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/28/869628/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-rally","ref":"refs/changes/28/869628/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-rally refs/changes/28/869628/3"}}},"commit":{"parents":[{"commit":"a179d50c7468ad7f9c871db499c872b4805c73ba","subject":"Update master for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/a179d50c7468ad7f9c871db499c872b4805c73ba"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-10 04:44:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-01-10 04:52:50.000000000","tz":540},"subject":"Replace outdated command to create rally db","message":"Replace outdated command to create rally db\n\nThe rally-manage db command was already replaced by the rally db\ncommand[1]. This fixes the outdated command still used.\n\n[1] https://review.opendev.org/c/openstack/rally/+/681224\n\nCloses-Bug: #2002372\nChange-Id: I34570e7c4131c3f06a1da23e9f2e57eec75ff25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/16a35f2fc6fdcaf039f7c483449b9b9230324659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-rally/commit/16a35f2fc6fdcaf039f7c483449b9b9230324659"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
