)]}'
{"id":"openstack%2Fpuppet-sahara~900608","triplet_id":"openstack%2Fpuppet-sahara~master~I224a5e32bd20aa86e14d738efd80d88ff9505a39","project":"openstack/puppet-sahara","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-15 16:46:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I224a5e32bd20aa86e14d738efd80d88ff9505a39","subject":"Add defined type to manage plugin packages","status":"MERGED","created":"2023-11-10 11:55:41.000000000","updated":"2023-11-15 16:47:57.000000000","submitted":"2023-11-15 16:46:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900608","meta_rev_id":"cc6b31ff0872b895e2e4609625acb7446a628eae","_number":900608,"virtual_id_number":900608,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-15 16:46:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-11-13 09:11:02.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2023-11-15 15:43:23.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2023-11-15 15:43:23.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-10 13:02:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 09:11:02.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-11-15 15:43:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ac015d3ae41bf38a23719a293b2ae9bed3bdfe93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-10 11:55:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"034b0bf27f4abbf3ee527cd3f8498f835f6e5e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 13:02:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67dd69f3addd4676985cc991b14cd5ec\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/730d8bf0dc0849dcb7fa7fa91974b40e : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dd9a05e95c740dca37fc4954fac9238 : FAILURE in 2m 30s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/041c60e5623f44a48ee5183eec2c8b8c : FAILURE in 6m 23s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/435dc14cad7347f2b02e4a15d4ac875a : FAILURE in 11m 28s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43809a93bc07439e968e7c3d6763f650 : FAILURE in 7m 38s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/994ad53dde614767ae26f9a8dc297a05 : FAILURE in 9m 49s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cc42b2d5fa9473983bc9698bb030419 : FAILURE in 8m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0609998835124e5588269ba5986ee6b4 : SUCCESS in 42m 28s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb786a7c87c14898a1776ebc5efedecf : SUCCESS in 1h 01m 01s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/150376c1761d4026a50f8ff593022e8a : SUCCESS in 18m 25s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34575f0f679a43939ff144e66edb77a5 : SUCCESS in 16m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/117c04bd36c048b58371be228b51b134 : SUCCESS in 15m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27f6f36b5dca4261bd43ba679ad03ec2 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7e58faf2301286b9b36e627c3961e7730774500f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-10 13:31:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"26463f7a5d886479d379769f84deef2e4d513bb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:40:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66dc16da64b34a159e2995cc7338c8f0\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c50dfff7b46b46fab8d55d13f5fbff45 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c3b4e4ffc854cd39eb4bbd69721fefb : SUCCESS in 2m 38s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/946fbff7375d4cfa86907bfccf135858 : SUCCESS in 5m 47s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/27e8dd39a0e94745bcb9edc8479911d6 : SUCCESS in 10m 11s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2418ba7719814141877a2091f5f2a9de : SUCCESS in 6m 42s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/4bf94cc518dd4baaaee8f500b9a15841 : SUCCESS in 10m 14s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6df84c3a36434145a5b5e2a2fe19e9a7 : SUCCESS in 6m 38s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa3d92aef5b445969598327befa1d5db : SUCCESS in 41m 39s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/be395ebfc3054653b5b95032c87aedf3 : SUCCESS in 1h 02m 24s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/36282287e8464c7eb21e993c40a3c755 : SUCCESS in 17m 48s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b57c596811a745b19fa0044e551eaf3c : SUCCESS in 15m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4943bf1c340144e69dc63b3baa7de213 : SUCCESS in 12m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ca531fab57d4efea2d693ea56b43194 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"12c38216396df0abbc825c886bd2f116faf0184e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-11-13 09:11:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5677f479f04865d37a1a86646e92304f108ce063","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-11-15 15:43:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b55fb366cda42ac0ccf744ed9d64f8e908ec89cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 15:44:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"be072fa41d7c1d88c921b5518b58742478d6f2a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 16:46:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d65af0e9ae934dc0a7513763b1b7cbf6\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4c9eaebc283341849bf388f51f7c1efb : SUCCESS in 3m 12s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db7104da68044d73b9c2fc4706e1c56b : SUCCESS in 2m 57s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/54b4b85ce3c148518408857662496542 : SUCCESS in 10m 40s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1010c6c1f19b415794461a5f56f7ff29 : SUCCESS in 7m 05s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1ac707bb64143aa9d4c52d79be29084 : SUCCESS in 10m 28s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfb5c43eb3ae4b9e955e3080f6c09002 : SUCCESS in 1h 00m 49s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b7592725d89420eae02602626dcdd9d : SUCCESS in 19m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/102d28eafab34a3183f8538944d3ead6 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":2},{"id":"c721169522ef68c5c9c02580e1a596524c3285cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 16:46:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cc6b31ff0872b895e2e4609625acb7446a628eae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 16:47:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e379c029ce154fec89f009a51f72ad73\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e52675b05eef4f349193da9488c0a4bf : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1445e3aca7722b2a3c630e041a7226cfaae2a2a1","revisions":{"bf4ab1fcba0f2521266be3a9733fbafd74f39e0a":{"kind":"REWORK","_number":1,"created":"2023-11-10 11:55:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/900608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-sahara","ref":"refs/changes/08/900608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-sahara refs/changes/08/900608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-sahara refs/changes/08/900608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-sahara refs/changes/08/900608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-sahara refs/changes/08/900608/1"}}},"commit":{"parents":[{"commit":"743331c9f475c7ed721eee0c8842c3a363c0a384","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-sahara/commit/743331c9f475c7ed721eee0c8842c3a363c0a384"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-10 11:54:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-10 11:54:23.000000000","tz":540},"subject":"Add defined type to manage plugin packages","message":"Add defined type to manage plugin packages\n\nSahara has separate plugin packages according to the software used in\nthe data processing cluster, and in Ubuntu we have to install\nthe required plugin packages additionally.\n\nThis introduces the defined resource type which can be used to manage\nplugin packages.\n\nChange-Id: I224a5e32bd20aa86e14d738efd80d88ff9505a39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-sahara/commit/bf4ab1fcba0f2521266be3a9733fbafd74f39e0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-sahara/commit/bf4ab1fcba0f2521266be3a9733fbafd74f39e0a"}]},"branch":"refs/heads/master"},"1445e3aca7722b2a3c630e041a7226cfaae2a2a1":{"kind":"REWORK","_number":2,"created":"2023-11-10 13:31:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/900608/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-sahara","ref":"refs/changes/08/900608/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-sahara refs/changes/08/900608/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-sahara refs/changes/08/900608/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-sahara refs/changes/08/900608/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-sahara refs/changes/08/900608/2"}}},"commit":{"parents":[{"commit":"743331c9f475c7ed721eee0c8842c3a363c0a384","subject":"Remove usage of deleted manifest_dir","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-sahara/commit/743331c9f475c7ed721eee0c8842c3a363c0a384"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-10 11:54:23.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-10 13:31:00.000000000","tz":540},"subject":"Add defined type to manage plugin packages","message":"Add defined type to manage plugin packages\n\nSahara has separate plugin packages according to the software used in\nthe data processing cluster, and in Ubuntu we have to install\nthe required plugin packages additionally.\n\nThis introduces the defined resource type which can be used to manage\nplugin packages.\n\nChange-Id: I224a5e32bd20aa86e14d738efd80d88ff9505a39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-sahara/commit/1445e3aca7722b2a3c630e041a7226cfaae2a2a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-sahara/commit/1445e3aca7722b2a3c630e041a7226cfaae2a2a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
