)]}'
{"id":"openstack%2Fpuppet-senlin~654206","triplet_id":"openstack%2Fpuppet-senlin~master~I64e98a796ec7d702b6bf5794eaf33210b41ef8cb","project":"openstack/puppet-senlin","branch":"master","topic":"move-to-opendev","hashtags":[],"change_id":"I64e98a796ec7d702b6bf5794eaf33210b41ef8cb","subject":"Move to opendev","status":"MERGED","created":"2019-04-21 06:46:03.000000000","updated":"2019-04-22 17:30:16.000000000","submitted":"2019-04-22 17:30:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654206-1555954216303-7a5aba1f","meta_rev_id":"e37735759449f879ca8dd1e8742d98f54b2c4dd7","_number":654206,"virtual_id_number":654206,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-22 17:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2019-04-22 13:25:35.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-22 17:06:20.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-04-22 17:06:20.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 13:25:35.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-22 17:06:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-22 17:30:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3fa0752431a517f77ab2f38164b99ee61c23f48","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 06:46:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29c9fd98b8c6f63c775cdf93ca46140bbd68572f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 07:28:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/06/654206/1/check/puppet-openstack-lint-ubuntu-bionic/ca21db2/ : SUCCESS in 4m 36s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/06/654206/1/check/puppet-openstack-syntax-4-ubuntu-bionic/311887b/ : SUCCESS in 4m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/06/654206/1/check/puppet-openstack-syntax-5-ubuntu-bionic/281b3c2/ : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/06/654206/1/check/puppet-openstack-syntax-6-ubuntu-bionic/647e1b6/ : SUCCESS in 5m 58s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/06/654206/1/check/puppet-openstack-unit-4.8-centos-7/6aae658/ : SUCCESS in 15m 10s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/06/654206/1/check/puppet-openstack-unit-5.5-centos-7/62a2cc5/ : SUCCESS in 15m 26s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/06/654206/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/bbd1d8f/ : SUCCESS in 7m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/06/654206/1/check/puppet-openstack-unit-latest-ubuntu-bionic/52365ab/ : SUCCESS in 6m 43s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/06/654206/1/check/puppet-openstack-beaker-centos-7/83a21f1/ : SUCCESS in 8m 58s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/06/654206/1/check/puppet-openstack-beaker-ubuntu-bionic/6bcbc21/ : SUCCESS in 8m 43s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/06/654206/1/check/build-openstack-releasenotes/8d21845/html/ : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":1},{"id":"421d68657de437931a4931ab3d8b367337b3a4dc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-22 13:25:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6b83d98636aa7c9795e48e8ce855c1210ee77a79","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-22 17:06:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c49cd97539ce109b32b16714ccf699047c933d81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:06:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb7df7d1ecea39c5174773a6f728d705d3200c4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:30:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"39320f6d69bde641fba7288193ebcfc9c9da56c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:30:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/06/654206/1/gate/puppet-openstack-lint-ubuntu-bionic/350d58b/ : SUCCESS in 6m 47s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/06/654206/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/5700f9f/ : SUCCESS in 6m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/06/654206/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/97affa0/ : SUCCESS in 7m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/06/654206/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/2155912/ : SUCCESS in 6m 57s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/06/654206/1/gate/puppet-openstack-unit-4.8-centos-7/b3392a6/ : SUCCESS in 9m 09s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/06/654206/1/gate/puppet-openstack-unit-5.5-centos-7/469ad84/ : SUCCESS in 9m 36s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/06/654206/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/bd379e5/ : SUCCESS in 5m 52s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/06/654206/1/gate/puppet-openstack-beaker-centos-7/8f14d49/ : SUCCESS in 9m 57s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/06/654206/1/gate/puppet-openstack-beaker-ubuntu-bionic/dea037f/ : SUCCESS in 5m 18s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/06/654206/1/gate/build-openstack-releasenotes/0d0d463/html/ : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15185d4194b1613055e5bef131f258d8706704cf","revisions":{"15185d4194b1613055e5bef131f258d8706704cf":{"kind":"REWORK","_number":1,"created":"2019-04-21 06:46:03.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/06/654206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-senlin","ref":"refs/changes/06/654206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/06/654206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/06/654206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/06/654206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-senlin refs/changes/06/654206/1"}}},"commit":{"parents":[{"commit":"5f5eddf089d9de5a65c4d859d7f4cdeac4bd86e8","subject":"Fix Repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/5f5eddf089d9de5a65c4d859d7f4cdeac4bd86e8"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:49.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: I64e98a796ec7d702b6bf5794eaf33210b41ef8cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/15185d4194b1613055e5bef131f258d8706704cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/15185d4194b1613055e5bef131f258d8706704cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
