)]}'
{"id":"openstack%2Fpuppet-senlin~665815","triplet_id":"openstack%2Fpuppet-senlin~master~I3fc4ee7a7a6761ee8a1ce2e5afc9a94b96c2a21e","project":"openstack/puppet-senlin","branch":"master","topic":"665815","hashtags":[],"change_id":"I3fc4ee7a7a6761ee8a1ce2e5afc9a94b96c2a21e","subject":"assert_private() params and deps classes","status":"ABANDONED","created":"2019-06-17 23:35:53.000000000","updated":"2020-09-18 22:57:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1bec4413006e048c35a8f73e351d3309fb874ab7","_number":665815,"virtual_id_number":665815,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-06-18 00:38:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 00:38:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0608308a0db50ec9b46abd620631e6b2a7c0afc2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:35:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1755ee08e7d46bd5b28de3135ee48832db752005","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:53:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"efcdc1e62e51bdbbb6e025991e07fbd7267639ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 00:38:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/15/665815/2/check/puppet-openstack-lint-ubuntu-bionic/90f5531/ : SUCCESS in 7m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/15/665815/2/check/puppet-openstack-syntax-5-ubuntu-bionic/632b7c9/ : SUCCESS in 6m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/15/665815/2/check/puppet-openstack-syntax-6-ubuntu-bionic/3e850b1/ : SUCCESS in 6m 45s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/15/665815/2/check/puppet-openstack-unit-5.5-centos-7/c09d6da/ : SUCCESS in 10m 19s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/15/665815/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/a246091/ : SUCCESS in 8m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/15/665815/2/check/puppet-openstack-unit-latest-ubuntu-bionic/d23afb9/ : SUCCESS in 9m 15s (non-voting)\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/15/665815/2/check/puppet-openstack-beaker-centos-7/3af3fc0/ : SUCCESS in 9m 58s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/15/665815/2/check/puppet-openstack-beaker-ubuntu-bionic/f55f0cf/ : SUCCESS in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b4f2c7fd0980e4aecae9d4db13739329b955b465","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-18 22:57:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"805e604b8903704246cdcd9ac24bfe81c7e91649","revisions":{"d204843f0efe1ac66feb1308bfbb587ef45c98cb":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:35:53.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/15/665815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-senlin","ref":"refs/changes/15/665815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/15/665815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/15/665815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/15/665815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-senlin refs/changes/15/665815/1"}}},"commit":{"parents":[{"commit":"040ca752392ab69a804cfef204898ac00de78ab1","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/040ca752392ab69a804cfef204898ac00de78ab1"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:35:45.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:35:45.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I3fc4ee7a7a6761ee8a1ce2e5afc9a94b96c2a21e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/d204843f0efe1ac66feb1308bfbb587ef45c98cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/d204843f0efe1ac66feb1308bfbb587ef45c98cb"}]},"branch":"refs/heads/master"},"805e604b8903704246cdcd9ac24bfe81c7e91649":{"kind":"REWORK","_number":2,"created":"2019-06-17 23:53:07.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/15/665815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-senlin","ref":"refs/changes/15/665815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/15/665815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/15/665815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/15/665815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-senlin refs/changes/15/665815/2"}}},"commit":{"parents":[{"commit":"040ca752392ab69a804cfef204898ac00de78ab1","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/040ca752392ab69a804cfef204898ac00de78ab1"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:35:45.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:53:01.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I3fc4ee7a7a6761ee8a1ce2e5afc9a94b96c2a21e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/805e604b8903704246cdcd9ac24bfe81c7e91649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/805e604b8903704246cdcd9ac24bfe81c7e91649"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
