)]}'
{"id":"openstack%2Fpuppet-senlin~676170","triplet_id":"openstack%2Fpuppet-senlin~master~I7aa34aa07731a509a5cec150bbbfa890b299e086","project":"openstack/puppet-senlin","branch":"master","topic":"dev","hashtags":[],"change_id":"I7aa34aa07731a509a5cec150bbbfa890b299e086","subject":"Add bugs url link to README","status":"ABANDONED","created":"2019-08-13 11:52:42.000000000","updated":"2019-08-15 08:31:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3bec315cfd5cd825b612eb5fcc1446dc02a10565","_number":676170,"virtual_id_number":676170,"owner":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-08-13 12:00:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":-1,"date":"2019-08-15 08:05:29.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 12:00:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 08:05:29.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"d5c330a6d2cea73a95e482e938eabfcd4e0fd6ca","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-08-13 11:52:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"552ad2d8d781a7adc3ac92da4fec8d53420d6e11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 12:00:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/941a38c379424041b3daaccf7746965f : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"5c72f3da8bf9249062477e3819c2e754ac8c23c2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-15 08:05:29.000000000","message":"Patch Set 1: Code-Review-1\n\nWe shouldn\u0027t do this, we are planning to move to StoryBoard which would require us to change these all over again.","accounts_in_message":[],"_revision_number":1},{"id":"a129a35791519f1e9c65edd9d6423151af4c5c43","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-08-15 08:31:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7f1cbd5669146594bf6a196ae33e2f5de050993","revisions":{"e7f1cbd5669146594bf6a196ae33e2f5de050993":{"kind":"REWORK","_number":1,"created":"2019-08-13 11:52:42.000000000","uploader":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"ref":"refs/changes/70/676170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-senlin","ref":"refs/changes/70/676170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/70/676170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/70/676170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/70/676170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-senlin refs/changes/70/676170/1"}}},"commit":{"parents":[{"commit":"a8a92d505ecdca7ed9ca0358162558dc6f1aab37","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/a8a92d505ecdca7ed9ca0358162558dc6f1aab37"}]}],"author":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-08-13 11:44:03.000000000","tz":-420},"committer":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-08-13 11:52:37.000000000","tz":-420},"subject":"Add bugs url link to README","message":"Add bugs url link to README\n\nChange-Id: I7aa34aa07731a509a5cec150bbbfa890b299e086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/e7f1cbd5669146594bf6a196ae33e2f5de050993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/e7f1cbd5669146594bf6a196ae33e2f5de050993"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
