)]}'
{"id":"openstack%2Fpuppet-senlin~728204","triplet_id":"openstack%2Fpuppet-senlin~stable%2Fussuri~I822cd4dd20d8e45a4c4b3dab3a6636f9650a32ed","project":"openstack/puppet-senlin","branch":"stable/ussuri","hashtags":[],"change_id":"I822cd4dd20d8e45a4c4b3dab3a6636f9650a32ed","subject":"Switch to Ussuri","status":"MERGED","created":"2020-05-14 18:09:20.000000000","updated":"2020-05-18 06:24:32.000000000","submitted":"2020-05-18 06:24:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728204-1589783072566-be6e565f","meta_rev_id":"2bf27afc27428eeb29466b6226e30c78d674c280","_number":728204,"virtual_id_number":728204,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 06:24:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-05-18 06:00:00.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-05-15 22:50:22.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-05-18 06:00:00.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 22:50:22.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-18 06:00:00.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-05-18 06:24:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce86d21e96bce3169c59c13b86a307f0345be25b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 18:09:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4aa818a275e320ff7c8556ce05b32318f3e7f4bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:42:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20f6369987c3442086d0114e3a5b4789 : SUCCESS in 5m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7a0efccec334184a471e0e1452cd2ac : SUCCESS in 5m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8d67da01ff44e009be84c4cc601735d : SUCCESS in 4m 38s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5c54f6f2d6544c49a020bd455701818b : SUCCESS in 16m 08s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/3c61140dc8d5459182d074f681db1e00 : SUCCESS in 12m 36s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a0dd667ed7241c98cc672c616dfbc82 : SUCCESS in 9m 10s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89b934720b7248dd8789af52e15ed753 : SUCCESS in 12m 58s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7f088342d5a1408c9fcc1a6806b375c7 : SUCCESS in 13m 30s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/f87a90fdaeec432ba5e5adba742fed52 : SUCCESS in 9m 10s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d2f293861dc4dc6a388ac8048f22cbe : SUCCESS in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ebf81723d2c56177a509c4e8f994510f7e34b368","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-15 22:50:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"38e358df865454a85bde9a991ac2fad97767d340","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-05-18 06:00:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d8d896cc02aa0c0324696fd6b7a7d3f7e5283b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 06:00:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a39291f653ca33a07ac249e152642eebf3c8a73b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 06:24:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2bf27afc27428eeb29466b6226e30c78d674c280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 06:24:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a505f41eaca449e5ad61aacd2ba88fe1 : SUCCESS in 6m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/306568787cd446499e213c35e3226c4d : SUCCESS in 7m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c45e793448e4b5fa86da4bccc1cd702 : SUCCESS in 7m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3811d89d66f0477bbcfc957d14a91083 : SUCCESS in 14m 22s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/f9d0afea7b1246c39281547fc8633620 : SUCCESS in 12m 42s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90ce4fa48aae45dc868308b6404029a0 : SUCCESS in 12m 02s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1a7e8387d63546a2a7835b6d73dd8f31 : SUCCESS in 11m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3e2e66d749390de7fe023cd8d9a3d617e6c481e0","revisions":{"3e2e66d749390de7fe023cd8d9a3d617e6c481e0":{"kind":"REWORK","_number":1,"created":"2020-05-14 18:09:20.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/04/728204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-senlin","ref":"refs/changes/04/728204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/04/728204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/04/728204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/04/728204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-senlin refs/changes/04/728204/1"}}},"commit":{"parents":[{"commit":"cfcb9944df98ea607b3ab71ce2def467911ed146","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/cfcb9944df98ea607b3ab71ce2def467911ed146"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 18:09:12.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 18:09:12.000000000","tz":120},"subject":"Switch to Ussuri","message":"Switch to Ussuri\n\nChange-Id: I822cd4dd20d8e45a4c4b3dab3a6636f9650a32ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/3e2e66d749390de7fe023cd8d9a3d617e6c481e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/3e2e66d749390de7fe023cd8d9a3d617e6c481e0"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
