)]}'
{"id":"openstack%2Fpuppet-senlin~729423","triplet_id":"openstack%2Fpuppet-senlin~master~I9024b06e50549f2648acc210b4addd9f9629f934","project":"openstack/puppet-senlin","branch":"master","topic":"db-password","hashtags":[],"change_id":"I9024b06e50549f2648acc210b4addd9f9629f934","subject":"Remove password hash generation in each puppet modules","status":"MERGED","created":"2020-05-19 23:24:31.000000000","updated":"2020-05-21 17:13:19.000000000","submitted":"2020-05-21 17:13:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729423-1590081199943-7d9b3224","meta_rev_id":"c10155433502a7419ba2e4e5b787bc0a43de20c5","_number":729423,"virtual_id_number":729423,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 17:13:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2020-05-20 08:35:10.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-05-21 16:49:15.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-05-21 16:49:15.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 08:35:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-21 16:49:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-21 17:13:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b9c211d5f494c366a94b37bc292d3570a0c530d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-19 23:24:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca3487dbd09dec2af38c4ad7728ca8bd14b841a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 23:45:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/8d414e0626af487daf456daf42bd1616 : SUCCESS in 2m 44s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc5d041e8bfe45d5baa8be8ba7a5cdac : SUCCESS in 4m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64fa7eedd69f44d5b4e49d2af5e7d5ee : SUCCESS in 5m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27f2c84109564395923ef89ec8245033 : SUCCESS in 4m 28s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/adb36f408b96407495e4b99bbd487b0a : FAILURE in 14m 02s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1fcd88ee368043cb8f18622c85e81f5b : FAILURE in 14m 53s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/548304bc84d44263877c1002cb8887b9 : FAILURE in 8m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d21efcd6bcc4c0ab04a78bc710ba465 : FAILURE in 8m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1bde4d63bb9c45bb98762fdc1bac3a81 : SUCCESS in 12m 00s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/925106ca59734185a2b9c1194a47551b : SUCCESS in 8m 00s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09f2e5099beb47d5a62b3b62e6c1bc23 : SUCCESS in 6m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"47c4390af9b9cf854b7d3d931803cbd04aaf7e29","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-20 00:04:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e26d8e85f9378dc8927daef16b3c706180b217aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 00:20:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/bdb352afa2fe4901bab101f85098506d : SUCCESS in 2m 44s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a12dbe0ee4f744f095fd9c4a1692c07e : SUCCESS in 7m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/595fa98025994f24a5070f2d268b264d : SUCCESS in 5m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/565bb771895846039535f97d2d54b3e2 : SUCCESS in 4m 55s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ed09b53dec714e6dbb28a3f2c4e67858 : SUCCESS in 9m 56s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/72baeb5153004cadab1d7a768b7f3fcf : SUCCESS in 10m 54s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7037668f56d49c6b682439a14c1a8e8 : SUCCESS in 8m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4616a50368af43819ddba0bd284cc694 : SUCCESS in 8m 21s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/331a025a2d5b4f109478af2305788378 : SUCCESS in 8m 34s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/85934e764d4340068888ca880481c0a8 : SUCCESS in 12m 26s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/720b5f1e230e4f94bea293629c994b74 : SUCCESS in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b67852ce126319a7f4334086ba2ed70df94fac3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-20 08:35:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"636061936a78b325ae797488b0a939a95b5fbdb7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-21 16:49:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5936a377d8cf9da39e63089bc6702a2ba8dd0baa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 16:51:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"65b8f19b82ebfe1261418b8a70b070d4d9109966","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 17:13:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/1fe4f032cd844345901fd5c992c9d301 : SUCCESS in 3m 15s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e913fb25572242beb00384ecfdc24472 : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/088ee948f73b49e1b578de1ec9e3cbf8 : SUCCESS in 6m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35fad4241bce4da3b16d7be95a91a0a3 : SUCCESS in 5m 49s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/66689b7a59cb49c9aa007c5854a47f93 : SUCCESS in 15m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b4ab39b04d0f45479e0fcda80938f8de : SUCCESS in 16m 03s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b972cb8b04d4ed0a060b6f91f9d0cf7 : SUCCESS in 17m 13s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/01a335c7ddaa4d21a817a2054283c697 : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3986a462ef5e3e970f0295ab06d81e16b0def50e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 17:13:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"77be720c4efbf9612fd3b44b9e98936c15dcf1a6","revisions":{"14e497d2e7af3abb0c6a4d0f459806645ff07c18":{"kind":"REWORK","_number":1,"created":"2020-05-19 23:24:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/729423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-senlin","ref":"refs/changes/23/729423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/23/729423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/23/729423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/23/729423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-senlin refs/changes/23/729423/1"}}},"commit":{"parents":[{"commit":"cc4631f4ede11e805cb0e591431c961fc88cc68d","subject":"Merge \"[ussuri][goal] Add CONTRIBUTING.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/cc4631f4ede11e805cb0e591431c961fc88cc68d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 23:24:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 23:24:19.000000000","tz":540},"subject":"Remove password hash generation in each puppet modules","message":"Remove password hash generation in each puppet modules\n\n... and migrate it to openstacklib so that all logics about database\nconfiguration are implemented in one common place.\n\nDepends-on: https://review.opendev.org/#/c/728595/\nChange-Id: I9024b06e50549f2648acc210b4addd9f9629f934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/14e497d2e7af3abb0c6a4d0f459806645ff07c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/14e497d2e7af3abb0c6a4d0f459806645ff07c18"}]},"branch":"refs/heads/master"},"77be720c4efbf9612fd3b44b9e98936c15dcf1a6":{"kind":"REWORK","_number":2,"created":"2020-05-20 00:04:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/729423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-senlin","ref":"refs/changes/23/729423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/23/729423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/23/729423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/23/729423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-senlin refs/changes/23/729423/2"}}},"commit":{"parents":[{"commit":"cc4631f4ede11e805cb0e591431c961fc88cc68d","subject":"Merge \"[ussuri][goal] Add CONTRIBUTING.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/cc4631f4ede11e805cb0e591431c961fc88cc68d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-19 23:24:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-20 00:04:45.000000000","tz":540},"subject":"Remove password hash generation in each puppet modules","message":"Remove password hash generation in each puppet modules\n\n... and migrate it to openstacklib so that all logics about database\nconfiguration are implemented in one common place.\n\nDepends-on: https://review.opendev.org/#/c/728595/\nChange-Id: I9024b06e50549f2648acc210b4addd9f9629f934\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/77be720c4efbf9612fd3b44b9e98936c15dcf1a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/77be720c4efbf9612fd3b44b9e98936c15dcf1a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
