)]}'
{"id":"openstack%2Fpuppet-senlin~747078","triplet_id":"openstack%2Fpuppet-senlin~master~I2151e4d19ae73b246ccd32189ff800c339cd10c1","project":"openstack/puppet-senlin","branch":"master","topic":"bug/1892284","hashtags":[],"change_id":"I2151e4d19ae73b246ccd32189ff800c339cd10c1","subject":"Add service_token_roles for keystone authtoken config","status":"MERGED","created":"2020-08-20 02:42:47.000000000","updated":"2020-08-20 21:36:14.000000000","submitted":"2020-08-20 21:34:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747078-1597959291091-7da3442f","meta_rev_id":"fbffda7c782f23571720a47fd100613a1d1aa474","_number":747078,"virtual_id_number":747078,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2020-08-20 20:52:17.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-20 21:34:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-20 12:11:32.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-20 14:36:38.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-20 14:36:38.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-20 12:11:32.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-20 14:36:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-08-20 20:52:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-20 21:34:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ff6d043404810ad4e2ff538d2e87b0b00db9beb","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-08-20 02:42:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"895fbeec34aba7b7cf4ecfc086a6bf3b21ed081f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 03:05:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf467b6011fc4b2281409e3ab9fa888c : SUCCESS in 4m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f58a3ec8132d40dcbe52a85b9fce286a : SUCCESS in 4m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b0c2bd252aa4f3c90c255f72fa42925 : SUCCESS in 4m 22s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7d988960683640de9618b54a67574cba : SUCCESS in 12m 14s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/bc2d17b564304082ae13d842f765826d : SUCCESS in 10m 06s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16a45597e2a3489fa917b6021b4e8088 : SUCCESS in 7m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9be36f43b233472483b30c1b49666f99 : FAILURE in 7m 13s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2328a8520e004552ad04e728e2b99cbc : SUCCESS in 8m 36s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/7d36e221764241ddb97b93e030bbcb95 : SUCCESS in 8m 25s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cff9d9f0c2a746368999f26983db8fbd : SUCCESS in 6m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53c8dfeb6d06497c8b78c4d88cbac11b : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"af1c071509f604cee4bd696edbd83ef271baa400","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-20 12:11:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8ddda46f3b729c6b89a8758efbb816f880023ee9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-20 14:36:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3906d7d27988268c0471eb6541d1827caea81f0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 14:47:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b66d5a8486698867a05f0c496f397dcbc2f8e40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 15:24:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fec60cdcd97c4a92a05d1dd9b0889c5a : RETRY_LIMIT in 1m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc36db8754694059bdc8c10ceb1dc4da : RETRY_LIMIT in 1m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7f0939191364dec87b11239443a50ea : RETRY_LIMIT in 1m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2ad2b07124804f87a29c9b7ee902e572 : SUCCESS in 12m 46s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/375ab5bf6c5e46d7af0834ae5910e045 : SUCCESS in 10m 05s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ca65911ce664d4397fe1d977733100c : RETRY_LIMIT in 1m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/85fe459b540844fbbbae7655d0dca159 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/837c670210ab4f99bc4a028f4ccfc605 : RETRY_LIMIT in 1m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8020028bfda159aff9601e64a4fc66adc1dda28f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-20 20:52:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5bd6985d3c85880daccc6fa5629bf1ae78280ba7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:16:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e11c99ce64b144d09ca3e63f60574b8d : SUCCESS in 5m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8cea2ae436d4347b3a3eae391092b96 : SUCCESS in 5m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86ed8c3f031d402197947d428fbb614a : SUCCESS in 5m 44s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/99c33dbf5fa545c199e642aecb0d7219 : SUCCESS in 10m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fc88dcd4e7224811b904cfd96922ef2e : SUCCESS in 9m 55s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5dc2d4323c4e4b10b4cf05c6ce60245b : SUCCESS in 8m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93943223344f42a2b1930a076f548ebb : FAILURE in 7m 41s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/28933beedcd9430087bd64b3f4d740ef : SUCCESS in 10m 43s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/d17f2a86d05f45989af593fd1bd51920 : SUCCESS in 9m 05s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecb73d36d3f4433ea53c723f792f4c89 : SUCCESS in 6m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff7e8ce3731449a5aa723d494dbf62fd : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9a7eb92b70aa18df2a7943b7106d73eeca3d7641","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:17:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f25b09d0a06408b14f8ff9521350ab0f221d650c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:34:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57049dbf519e4762b2f1c627b1e3827d : SUCCESS in 4m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/712b599195fc4af7ad25ab0260c74506 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78c97eb83e49433a9e041b6c0756c294 : SUCCESS in 4m 38s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b1f13fd93d634a03b23341da048602a5 : SUCCESS in 9m 20s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8390c2f26f1d4241b2ae82bd0d61b473 : SUCCESS in 10m 58s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d49c4cd3d6b84fc2afd1c713205ada2c : SUCCESS in 9m 22s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3d61231607ab4072a196caf856d39a5d : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/856ae7797fc74019bd884b3a0e3a7a6e : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"8753f9e039611525636735fcde9604e3eafbcaf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:34:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ac1e068908915d4e64f7bdb0e400c6ca3c41504a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:36:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/615f557f26bf4b06a5fff2882b6ccd53 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"222152a595781c3cdc5cddf5e1a31979f96149d7","revisions":{"222152a595781c3cdc5cddf5e1a31979f96149d7":{"kind":"REWORK","_number":1,"created":"2020-08-20 02:42:47.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/78/747078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-senlin","ref":"refs/changes/78/747078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/78/747078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/78/747078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/78/747078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-senlin refs/changes/78/747078/1"}}},"commit":{"parents":[{"commit":"8f33e2431d1e354587f5d956700aa6eea3399709","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/8f33e2431d1e354587f5d956700aa6eea3399709"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-20 02:41:27.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-08-20 02:41:27.000000000","tz":480},"subject":"Add service_token_roles for keystone authtoken config","message":"Add service_token_roles for keystone authtoken config\n\nAdd the ability to configure service_token_roles.\n\nChange-Id: I2151e4d19ae73b246ccd32189ff800c339cd10c1\nCloses-Bug: #1892284\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/222152a595781c3cdc5cddf5e1a31979f96149d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/222152a595781c3cdc5cddf5e1a31979f96149d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
