)]}'
{"id":"openstack%2Fpuppet-senlin~756831","triplet_id":"openstack%2Fpuppet-senlin~stable%2Fvictoria~Idd7205b0dca6f164104dc889f7b644fcf31cf026","project":"openstack/puppet-senlin","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"Idd7205b0dca6f164104dc889f7b644fcf31cf026","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-10-08 14:46:57.000000000","updated":"2020-10-09 19:33:11.000000000","submitted":"2020-10-09 19:33:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756831-1602271991085-c901a1f2","meta_rev_id":"71eb38e8fe312cae25bdce8863fb5ca2e0c67f51","_number":756831,"virtual_id_number":756831,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-09 19:33:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-08 14:54:24.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-10-09 18:47:40.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 14:54:24.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-09 18:47:40.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-10-09 19:33:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"973d458b1321307dd1515a371104e96e69641b41","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-10-08 14:46:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdf4e852bf2bad90d286d28ef63b910cad1664d0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-08 14:54:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa9c26e8c8947c87382b980dfddb45dc9e163bd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 15:41:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17bf1cb19b594a97b76640cb63834afe : SUCCESS in 3m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/502f96c8f5d149e9bfeb8b2bb8a78033 : SUCCESS in 3m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/34c259881b074167886017147e370898 : SUCCESS in 12m 48s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15971af935284319999691540cb6a4db : SUCCESS in 6m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c87dfedcc343483f93ef5aa4cfce071c : SUCCESS in 10m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65c728b7f4eea0a4ab678597a4d56ecb47a5b42d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-10-09 18:47:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb3ce2195b1529496a167bb9d2b1e7354f88c35d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 18:47:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76d4767e567719cf4904d86fe02943902686f25f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 19:33:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be75dd59c3f646f6a3843d81d1328440 : SUCCESS in 3m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c18735c5818444f690cdbf7e344de53f : SUCCESS in 3m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ae2a97c765084308bba1d4bc41d9a652 : SUCCESS in 12m 21s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e89f049eaab049a58ec346a528ea21c5 : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":1},{"id":"90486c4e63e5fb0487f3455c60c47a01d89c868a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 19:33:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af2abf7a53718832e8b787b4f01e8397b5fe5344","revisions":{"af2abf7a53718832e8b787b4f01e8397b5fe5344":{"kind":"REWORK","_number":1,"created":"2020-10-08 14:46:57.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/31/756831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-senlin","ref":"refs/changes/31/756831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/31/756831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/31/756831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-senlin refs/changes/31/756831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-senlin refs/changes/31/756831/1"}}},"commit":{"parents":[{"commit":"16fb2017c52bef29fa71adcc287b8824af5ec714","subject":"Prepare Victoria RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/16fb2017c52bef29fa71adcc287b8824af5ec714"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-08 14:46:55.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-08 14:46:55.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: Idd7205b0dca6f164104dc889f7b644fcf31cf026\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/af2abf7a53718832e8b787b4f01e8397b5fe5344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-senlin/commit/af2abf7a53718832e8b787b4f01e8397b5fe5344"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
