)]}'
{"id":"openstack%2Fpuppet-swift~419556","triplet_id":"openstack%2Fpuppet-swift~stable%2Fnewton~I67385d3a3f17f8baec657cc7f85153f9ff31bee4","project":"openstack/puppet-swift","branch":"stable/newton","topic":"CVE-2016-9590","hashtags":[],"change_id":"I67385d3a3f17f8baec657cc7f85153f9ff31bee4","subject":"[CVE-2016-9590] Correct configuration file perms","status":"MERGED","created":"2017-01-12 16:45:25.000000000","updated":"2017-01-12 19:01:43.000000000","submitted":"2017-01-12 19:01:43.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b79c641e100918413c0792785859ec46af112180","_number":419556,"virtual_id_number":419556,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2017-01-12 19:01:43.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2017-01-12 16:52:13.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-01-12 17:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2017-01-12 18:12:04.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-01-12 17:02:55.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2017-01-12 18:12:04.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-01-12 19:01:43.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5bc1bc37987f79174e704d2de4ae7774e7ce2768","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-01-12 16:45:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afd7f9da09c2d1354f03bdbb5cb6d57d5a5312ab","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-01-12 16:52:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6c051c2e0a7f9b2a44553202d11eef02fd8f34b","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2017-01-12 17:02:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"31881e83edad80dd31584276bde7f818514f0ad4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-12 17:59:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-puppet-swift-puppet-lint http://logs.openstack.org/56/419556/1/check/gate-puppet-swift-puppet-lint/49951d9/ : SUCCESS in 3m 41s\n- gate-puppet-swift-puppet-syntax-3-centos-7 http://logs.openstack.org/56/419556/1/check/gate-puppet-swift-puppet-syntax-3-centos-7/a8477fb/ : SUCCESS in 3m 28s\n- gate-puppet-swift-puppet-syntax-4-centos-7 http://logs.openstack.org/56/419556/1/check/gate-puppet-swift-puppet-syntax-4-centos-7/df929da/ : SUCCESS in 1m 57s\n- gate-puppet-swift-puppet-unit-3.6-centos-7 http://logs.openstack.org/56/419556/1/check/gate-puppet-swift-puppet-unit-3.6-centos-7/6a6fdb6/ : SUCCESS in 5m 58s\n- gate-puppet-swift-puppet-unit-3.8-centos-7 http://logs.openstack.org/56/419556/1/check/gate-puppet-swift-puppet-unit-3.8-centos-7/9774fb1/ : SUCCESS in 5m 44s\n- gate-puppet-swift-puppet-unit-4.5-centos-7 http://logs.openstack.org/56/419556/1/check/gate-puppet-swift-puppet-unit-4.5-centos-7/72dd498/ : SUCCESS in 7m 45s\n- gate-puppet-swift-puppet-beaker-rspec-centos-7 http://logs.openstack.org/56/419556/1/check/gate-puppet-swift-puppet-beaker-rspec-centos-7/d10856c/ : SUCCESS in 12m 01s\n- gate-puppet-swift-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/56/419556/1/check/gate-puppet-swift-puppet-beaker-rspec-ubuntu-xenial/c7c62b4/ : SUCCESS in 11m 51s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/56/419556/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/1f04919/ : SUCCESS in 37m 34s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/56/419556/1/check/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/85d4f79/ : SUCCESS in 27m 01s\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/56/419556/1/check/gate-tripleo-ci-centos-7-nonha-multinode/e80c91b/ : SUCCESS in 1h 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f554a34ca22627cf39aebea1130b3da8623e4638","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-01-12 18:12:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"650bbb5e4fec6250db08d6079ce486645b90d873","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-12 18:12:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"38ef593d4a427dc19fdbe640a0d45a928c0df574","author":{"_account_id":8971,"name":"Fuel CI","username":"fuel-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2017-01-12 18:51:30.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* newton.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron https://ci.fuel-infra.org/job/newton.puppet-openstack.fuel-library.pkgs.ubuntu.smoke_neutron/239/ : FAILURE\n\n* newton.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library https://ci.fuel-infra.org/job/newton.puppet-openstack.fuel-library.pkgs.ubuntu.review_in_fuel_library/241/ : SUCCESS\n\n* newton.puppet-openstack.fuel.noop https://ci.fuel-infra.org/job/newton.puppet-openstack.fuel.noop/241/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f355d723da4a71ed632e90fc268d50504e6b433a","date":"2017-01-12 19:01:43.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"4395acbe3f239d03cddeb781cb71a3c0c899298a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-12 19:01:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-puppet-swift-puppet-lint http://logs.openstack.org/56/419556/1/gate/gate-puppet-swift-puppet-lint/189e835/ : SUCCESS in 2m 07s\n- gate-puppet-swift-puppet-syntax-3-centos-7 http://logs.openstack.org/56/419556/1/gate/gate-puppet-swift-puppet-syntax-3-centos-7/8f2d34d/ : SUCCESS in 3m 20s\n- gate-puppet-swift-puppet-syntax-4-centos-7 http://logs.openstack.org/56/419556/1/gate/gate-puppet-swift-puppet-syntax-4-centos-7/c7076dd/ : SUCCESS in 2m 18s\n- gate-puppet-swift-puppet-unit-3.6-centos-7 http://logs.openstack.org/56/419556/1/gate/gate-puppet-swift-puppet-unit-3.6-centos-7/8df8b29/ : SUCCESS in 6m 11s\n- gate-puppet-swift-puppet-unit-3.8-centos-7 http://logs.openstack.org/56/419556/1/gate/gate-puppet-swift-puppet-unit-3.8-centos-7/3444f3c/ : SUCCESS in 6m 31s\n- gate-puppet-swift-puppet-unit-4.5-centos-7 http://logs.openstack.org/56/419556/1/gate/gate-puppet-swift-puppet-unit-4.5-centos-7/3b18dc5/ : SUCCESS in 6m 34s\n- gate-puppet-swift-puppet-beaker-rspec-centos-7 http://logs.openstack.org/56/419556/1/gate/gate-puppet-swift-puppet-beaker-rspec-centos-7/cccc833/ : SUCCESS in 17m 32s\n- gate-puppet-swift-puppet-beaker-rspec-ubuntu-xenial http://logs.openstack.org/56/419556/1/gate/gate-puppet-swift-puppet-beaker-rspec-ubuntu-xenial/84a2e76/ : SUCCESS in 10m 45s\n- gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial http://logs.openstack.org/56/419556/1/gate/gate-puppet-openstack-integration-4-scenario002-tempest-ubuntu-xenial/6e89c58/ : SUCCESS in 26m 45s\n- gate-puppet-openstack-integration-4-scenario002-tempest-centos-7 http://logs.openstack.org/56/419556/1/gate/gate-puppet-openstack-integration-4-scenario002-tempest-centos-7/624fc59/ : SUCCESS in 28m 38s\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/56/419556/1/gate/gate-tripleo-ci-centos-7-nonha-multinode/6832f65/ : SUCCESS in 49m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fff73b533b07e719b1daf65400dc47dcc84edf60","revisions":{"fff73b533b07e719b1daf65400dc47dcc84edf60":{"kind":"REWORK","_number":1,"created":"2017-01-12 16:45:25.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/56/419556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/56/419556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/56/419556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/56/419556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/56/419556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/56/419556/1"}}},"commit":{"parents":[{"commit":"04d4bf6de1cf30ecb036f25ab39aa4542b6d631f","subject":"set url_base option in staticweb swift middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/04d4bf6de1cf30ecb036f25ab39aa4542b6d631f"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2017-01-12 15:21:31.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2017-01-12 16:02:15.000000000","tz":-420},"subject":"[CVE-2016-9590] Correct configuration file perms","message":"[CVE-2016-9590] Correct configuration file perms\n\nThe configuration files for the swift proxy and storage server were\ngetting their permissions changed from 0640 to 0644 due to the missing\nmode on the concat resource for these files. This change defines the\npermissions for these files to be 0640 to prevent unauthorized reading\nof sensitive information that may be contained in these files.\n\nChange-Id: I67385d3a3f17f8baec657cc7f85153f9ff31bee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fff73b533b07e719b1daf65400dc47dcc84edf60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fff73b533b07e719b1daf65400dc47dcc84edf60"}]},"branch":"refs/heads/stable/newton"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
