)]}'
{"id":"openstack%2Fpuppet-swift~656481","triplet_id":"openstack%2Fpuppet-swift~master~I2fc7b23382c8cee1aec026cb669f2185de004171","project":"openstack/puppet-swift","branch":"master","topic":"ostempest","hashtags":[],"change_id":"I2fc7b23382c8cee1aec026cb669f2185de004171","subject":"Port puppet-swift-tripleo-standalone to os_tempest","status":"MERGED","created":"2019-04-30 09:24:17.000000000","updated":"2019-05-01 16:34:33.000000000","submitted":"2019-05-01 16:34:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656481-1556728473020-0411a076","meta_rev_id":"d5c16b3abfbc8f13df7952ce9513128c489b54f2","_number":656481,"virtual_id_number":656481,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-01 16:34:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-30 10:08:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":2,"date":"2019-05-01 14:45:04.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-30 14:04:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"date":"2019-04-30 09:24:17.000000000","_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"value":1,"date":"2019-05-01 14:45:04.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-04-30 12:28:20.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"date":"2019-04-30 12:29:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-30 09:25:03.000000000","_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-30 09:24:17.000000000","updated_by":{"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},"reviewer":{"_account_id":8944,"name":"Chandan Kumar","username":"Chandankumar","inactive":true},"state":"REVIEWER"},{"updated":"2019-04-30 09:25:03.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-04-30 10:08:58.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 12:28:20.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-04-30 12:29:21.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-30 14:04:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-01 14:45:04.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-05-01 16:34:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2b5a71c88179c6e040c90bd7d19a2b336f034b9b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-30 09:24:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"644245dbe24c3ae233eb0130174237453c23f798","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-30 10:08:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/81/656481/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/cce5ef5/ : SUCCESS in 37m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/81/656481/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/7c76a7c/ : SUCCESS in 40m 13s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/81/656481/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/4f367c7/ : SUCCESS in 40m 15s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/81/656481/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/9fa5140/ : FAILURE in 30m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dcbec822df17fc8f4b40ab48d39219c08f912d40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 11:00:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/81/656481/1/check/puppet-openstack-lint-ubuntu-bionic/4ec1f6a/ : SUCCESS in 7m 18s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/81/656481/1/check/puppet-openstack-syntax-4-ubuntu-bionic/b41eeb4/ : SUCCESS in 6m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/81/656481/1/check/puppet-openstack-syntax-5-ubuntu-bionic/47116a1/ : SUCCESS in 6m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/81/656481/1/check/puppet-openstack-syntax-6-ubuntu-bionic/471db2b/ : SUCCESS in 6m 47s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/81/656481/1/check/puppet-openstack-unit-4.8-centos-7/187aab7/ : SUCCESS in 21m 25s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/81/656481/1/check/puppet-openstack-unit-5.5-centos-7/a1c9ed2/ : SUCCESS in 18m 20s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/81/656481/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/4e6fa3d/ : SUCCESS in 12m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/81/656481/1/check/puppet-openstack-unit-latest-ubuntu-bionic/47471f6/ : SUCCESS in 14m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/81/656481/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/b0a4e10/ : FAILURE in 28m 08s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/81/656481/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/9b92dd7/ : SUCCESS in 57m 43s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/81/656481/1/check/puppet-openstack-beaker-centos-7/cef22bc/ : SUCCESS in 25m 11s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/81/656481/1/check/puppet-openstack-beaker-ubuntu-bionic/115cb8c/ : FAILURE in 14m 45s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/81/656481/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/d37eba6/ : SUCCESS in 1h 30m 37s (non-voting)\n- puppet-swift-tripleo-standalone http://logs.openstack.org/81/656481/1/check/puppet-swift-tripleo-standalone/65d20c2/ : SUCCESS in 1h 12m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b416ef1783d66282703c82488dd01592fd85559","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-30 14:04:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"867a137c4eafed12cd3a6814424ef6c8b794a1f7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-05-01 14:45:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b89d73aeb9bc6755eea6e36a6dd5ff40b78e9475","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 14:45:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0cd600a28fd1427edaec441b2efd7c14b0f93996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 16:34:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/81/656481/1/gate/puppet-openstack-lint-ubuntu-bionic/4645443/ : SUCCESS in 4m 34s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/81/656481/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/dbbd7e8/ : SUCCESS in 4m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/81/656481/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/f48097f/ : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/81/656481/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/c8acc36/ : SUCCESS in 4m 38s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/81/656481/1/gate/puppet-openstack-unit-4.8-centos-7/6caae5e/ : SUCCESS in 21m 00s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/81/656481/1/gate/puppet-openstack-unit-5.5-centos-7/5708457/ : SUCCESS in 18m 04s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/81/656481/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/0c9acc0/ : SUCCESS in 10m 41s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/81/656481/1/gate/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/88193af/ : FAILURE in 33m 03s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/81/656481/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/87125c7/ : SUCCESS in 54m 02s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/81/656481/1/gate/puppet-openstack-beaker-centos-7/5d8a873/ : SUCCESS in 22m 41s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/81/656481/1/gate/puppet-openstack-beaker-ubuntu-bionic/add8ca0/ : FAILURE in 14m 17s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/81/656481/1/gate/tripleo-puppet-ci-centos-7-undercloud-containers/0ad2460/ : SUCCESS in 1h 30m 41s (non-voting)\n- puppet-swift-tripleo-standalone http://logs.openstack.org/81/656481/1/gate/puppet-swift-tripleo-standalone/03f98d5/ : SUCCESS in 1h 12m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50da52b4dc3972c48d4d7a3ce7f5921b63496b3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 16:34:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9208f91a88085faaf93682198722683b0183dd03","revisions":{"9208f91a88085faaf93682198722683b0183dd03":{"kind":"REWORK","_number":1,"created":"2019-04-30 09:24:17.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/81/656481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/81/656481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/81/656481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/81/656481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/81/656481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/81/656481/1"}}},"commit":{"parents":[{"commit":"52d08aad6ce04289deb9f1308da2fb1de1399755","subject":"Merge \"Provide more useful fail message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/52d08aad6ce04289deb9f1308da2fb1de1399755"}]}],"author":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-30 09:23:25.000000000","tz":330},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-30 09:24:14.000000000","tz":330},"subject":"Port puppet-swift-tripleo-standalone to os_tempest","message":"Port puppet-swift-tripleo-standalone to os_tempest\n\nPreviously Tripleo validate-tempest role was used to run tempest\non tripleo jobs. Recently we are moving to os_tempest (an unified\nrole developed by openstack-ansible team).\n\nIt moves the vars to os_tempest.\nBelow is the other var changes:\nrun_tempest: false -\u003e to disable validate-tempest role\nuse_os_tempest: true -\u003e enable os_tempest\ntempest_whitelist -\u003e tempest_test_whitelist\n\nChange-Id: I2fc7b23382c8cee1aec026cb669f2185de004171\nSigned-off-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9208f91a88085faaf93682198722683b0183dd03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9208f91a88085faaf93682198722683b0183dd03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
