)]}'
{"id":"openstack%2Fpuppet-swift~660005","triplet_id":"openstack%2Fpuppet-swift~master~I58850e0a404bd83153b6c985c815115fc1b09f34","project":"openstack/puppet-swift","branch":"master","topic":"fix-client-test","hashtags":[],"change_id":"I58850e0a404bd83153b6c985c815115fc1b09f34","subject":"Fix client spec test","status":"MERGED","created":"2019-05-20 03:04:12.000000000","updated":"2019-05-20 07:18:19.000000000","submitted":"2019-05-20 07:18:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660005-1558336699708-354c625b","meta_rev_id":"6469dc17f772c9e39fae25184236cc59e22468bd","_number":660005,"virtual_id_number":660005,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-20 07:18:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-05-20 05:34:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-05-20 05:34:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-20 07:18:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0975e3a0c41f65ae476681fc264924eb771a95a2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 03:04:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbcefb60ec18bd531757a362428af5dfc203b96c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 04:46:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/05/660005/1/check/puppet-openstack-lint-ubuntu-bionic/7ba24ca/ : SUCCESS in 5m 23s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/05/660005/1/check/puppet-openstack-syntax-4-ubuntu-bionic/e2e4011/ : SUCCESS in 5m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/05/660005/1/check/puppet-openstack-syntax-5-ubuntu-bionic/9c8297c/ : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/05/660005/1/check/puppet-openstack-syntax-6-ubuntu-bionic/f393d3a/ : SUCCESS in 4m 28s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/05/660005/1/check/puppet-openstack-unit-4.8-centos-7/b8923e5/ : SUCCESS in 18m 23s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/05/660005/1/check/puppet-openstack-unit-5.5-centos-7/16cc2ca/ : SUCCESS in 16m 42s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/05/660005/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/112996b/ : SUCCESS in 13m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/05/660005/1/check/puppet-openstack-unit-latest-ubuntu-bionic/a830b14/ : SUCCESS in 12m 08s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/05/660005/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/c0b5156/ : SUCCESS in 1h 28m 11s (non-voting)\n- puppet-swift-tripleo-standalone http://logs.openstack.org/05/660005/1/check/puppet-swift-tripleo-standalone/a3a4fd5/ : SUCCESS in 1h 19m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1d4de974f055ffcf556cc7024b24a99c5855892","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-20 05:34:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"df06e339eb3a6cf6cabf1dff5606a8168c5b95f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 05:34:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b9d784b5db95ab5e0e354d0b21a0240d0da0c64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:18:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c9f825d777c7049700323651fa90a7ea6e4aa5bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 07:18:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/05/660005/1/gate/puppet-openstack-lint-ubuntu-bionic/26c25b0/ : SUCCESS in 4m 14s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/05/660005/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/275a054/ : SUCCESS in 3m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/05/660005/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/63d271f/ : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/05/660005/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/7f090d5/ : SUCCESS in 3m 47s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/05/660005/1/gate/puppet-openstack-unit-4.8-centos-7/2790e5e/ : SUCCESS in 19m 44s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/05/660005/1/gate/puppet-openstack-unit-5.5-centos-7/2656816/ : SUCCESS in 27m 12s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/05/660005/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/3e8d592/ : SUCCESS in 8m 57s\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/05/660005/1/gate/tripleo-puppet-ci-centos-7-undercloud-containers/e767b22/ : SUCCESS in 1h 42m 54s (non-voting)\n- puppet-swift-tripleo-standalone http://logs.openstack.org/05/660005/1/gate/puppet-swift-tripleo-standalone/d2d7511/ : SUCCESS in 1h 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fde777783a737af4c1bc67d035a5d514da35961","revisions":{"1fde777783a737af4c1bc67d035a5d514da35961":{"kind":"REWORK","_number":1,"created":"2019-05-20 03:04:12.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/05/660005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/05/660005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/05/660005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/05/660005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/05/660005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/05/660005/1"}}},"commit":{"parents":[{"commit":"9208f91a88085faaf93682198722683b0183dd03","subject":"Port puppet-swift-tripleo-standalone to os_tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9208f91a88085faaf93682198722683b0183dd03"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:21.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-20 03:03:21.000000000","tz":480},"subject":"Fix client spec test","message":"Fix client spec test\n\nChange-Id: I58850e0a404bd83153b6c985c815115fc1b09f34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1fde777783a737af4c1bc67d035a5d514da35961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1fde777783a737af4c1bc67d035a5d514da35961"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
