)]}'
{"id":"openstack%2Fpuppet-swift~661229","triplet_id":"openstack%2Fpuppet-swift~master~I6e11480a8d122a7b5ae9325357e558a0416f676c","project":"openstack/puppet-swift","branch":"master","topic":"memcached-inet6-prefix","hashtags":[],"change_id":"I6e11480a8d122a7b5ae9325357e558a0416f676c","subject":"Prefix memcached_servers IPv6 address with `inet6:`","status":"ABANDONED","created":"2019-05-24 08:04:35.000000000","updated":"2019-05-27 11:31:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2684b58d20be135d13371aec9b8d7908f3da4805","_number":661229,"virtual_id_number":661229,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-24 16:31:59.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":-1,"date":"2019-05-24 13:49:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-05-24 19:16:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"date":"2019-05-24 21:26:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 13:49:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-24 19:16:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-24 21:26:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"d6d1571a8c5ccf38a6b03b4d57efdede6e6d74e1","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-24 08:04:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c52f74651ad1debf99296a485e0b127f56ee36a8","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-24 08:13:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee9e2bd6dbb34d460d91045e4c875eb5a9de4d95","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-24 09:08:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/29/661229/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/098d2f6/ : SUCCESS in 52m 21s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/29/661229/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/82a588f/ : FAILURE in 9m 50s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/29/661229/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/25bbe29/ : SUCCESS in 49m 29s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/29/661229/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/bdee611/ : FAILURE in 42m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"10d134d5167ee9d02e6eaaa60747141d510dade3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 09:31:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/661229/2/check/puppet-openstack-lint-ubuntu-bionic/47194d1/ : SUCCESS in 5m 07s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/661229/2/check/puppet-openstack-syntax-4-ubuntu-bionic/15c3c25/ : SUCCESS in 4m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/661229/2/check/puppet-openstack-syntax-5-ubuntu-bionic/fbe0b4c/ : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/661229/2/check/puppet-openstack-syntax-6-ubuntu-bionic/5102b5d/ : SUCCESS in 6m 43s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/29/661229/2/check/puppet-openstack-unit-4.8-centos-7/1d08b9d/ : FAILURE in 14m 36s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/661229/2/check/puppet-openstack-unit-5.5-centos-7/83baf0e/ : FAILURE in 11m 41s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/29/661229/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/889b1db/ : FAILURE in 7m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/661229/2/check/puppet-openstack-unit-latest-ubuntu-bionic/b713f1b/ : FAILURE in 6m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/29/661229/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/a105624/ : FAILURE in 14m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/29/661229/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/9c14131/ : FAILURE in 7m 35s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/29/661229/2/check/puppet-openstack-beaker-centos-7/704c8dc/ : FAILURE in 9m 27s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/29/661229/2/check/puppet-openstack-beaker-ubuntu-bionic/2ff895e/ : FAILURE in 13m 36s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/29/661229/2/check/tripleo-puppet-ci-centos-7-undercloud-containers/493b899/ : FAILURE in 47m 49s (non-voting)\n- puppet-swift-tripleo-standalone http://logs.openstack.org/29/661229/2/check/puppet-swift-tripleo-standalone/5445e59/ : FAILURE in 45m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff7cafdd6712d15aead7a33ced4d139c21d8900d","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-24 09:39:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c72a8edcf2351b7f843d44ec197e437a63aca10f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 10:37:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/661229/3/check/puppet-openstack-lint-ubuntu-bionic/33f6b16/ : SUCCESS in 3m 51s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/661229/3/check/puppet-openstack-syntax-4-ubuntu-bionic/0e72312/ : SUCCESS in 3m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/661229/3/check/puppet-openstack-syntax-5-ubuntu-bionic/203c41d/ : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/661229/3/check/puppet-openstack-syntax-6-ubuntu-bionic/12eb9e1/ : SUCCESS in 5m 04s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/29/661229/3/check/puppet-openstack-unit-4.8-centos-7/9467dd5/ : FAILURE in 27m 39s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/661229/3/check/puppet-openstack-unit-5.5-centos-7/efdd867/ : FAILURE in 23m 04s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/29/661229/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/3484fa5/ : SUCCESS in 14m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/661229/3/check/puppet-openstack-unit-latest-ubuntu-bionic/3c1331b/ : SUCCESS in 14m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/29/661229/3/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/8017a65/ : FAILURE in 7m 17s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/29/661229/3/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/5df94fc/ : FAILURE in 8m 52s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/29/661229/3/check/puppet-openstack-beaker-centos-7/4c751a5/ : FAILURE in 15m 00s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/29/661229/3/check/puppet-openstack-beaker-ubuntu-bionic/a20afcd/ : FAILURE in 7m 47s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/29/661229/3/check/tripleo-puppet-ci-centos-7-undercloud-containers/2ab605c/ : FAILURE in 49m 45s (non-voting)\n- puppet-swift-tripleo-standalone http://logs.openstack.org/29/661229/3/check/puppet-swift-tripleo-standalone/352d7e6/ : FAILURE in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5c6bd5afeb0271fb17fe3e30d151dde54cc5c0d1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-24 11:20:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/29/661229/3/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/8fe896b/ : SUCCESS in 41m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/29/661229/3/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/04041d3/ : FAILURE in 6m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/29/661229/3/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/fbee504/ : SUCCESS in 43m 43s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/29/661229/3/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/717021f/ : FAILURE in 34m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09bc61fd00bffe605599d69761c6126c2851acb6","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-24 11:59:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"48705714c76e73d0cee7019193e31e72f377a3c9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-24 13:03:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/781697a/ : SUCCESS in 56m 54s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/27c4f71/ : FAILURE in 27m 38s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/acc8561/ : SUCCESS in 1h 00m 55s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/13d671d/ : SUCCESS in 50m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6d9144b4563923d5c357b89a2e981458efc385dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 13:49:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball http://logs.openstack.org/29/661229/4/check/build-openstack-puppet-tarball/45728ae/ : SUCCESS in 1m 34s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-lint-ubuntu-bionic/fb45c27/ : SUCCESS in 6m 34s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-syntax-4-ubuntu-bionic/634bcbb/ : SUCCESS in 4m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-syntax-5-ubuntu-bionic/aca7c0f/ : SUCCESS in 6m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-syntax-6-ubuntu-bionic/4004a66/ : SUCCESS in 3m 49s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-4.8-centos-7/fb11411/ : SUCCESS in 27m 37s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-5.5-centos-7/b015b45/ : SUCCESS in 24m 14s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-6.4-ubuntu-bionic/a21d2ef/ : SUCCESS in 16m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-latest-ubuntu-bionic/bf620f3/ : SUCCESS in 12m 29s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/439a61f/ : FAILURE in 32m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/b0e165d/ : FAILURE in 30m 05s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-beaker-centos-7/0120f2d/ : SUCCESS in 23m 04s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-beaker-ubuntu-bionic/4ecf035/ : FAILURE in 11m 59s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/29/661229/4/check/tripleo-puppet-ci-centos-7-undercloud-containers/e9985d3/ : SUCCESS in 1h 42m 26s (non-voting)\n- puppet-swift-tripleo-standalone http://logs.openstack.org/29/661229/4/check/puppet-swift-tripleo-standalone/10d3562/ : SUCCESS in 1h 08m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"03e9b008025eb4fca1b4da8bf474b379d9385ce3","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-24 14:37:06.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d19832c21b122706f198dda98e638b62f4c8ee2e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-24 15:26:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/dea83e3/ : SUCCESS in 47m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/bb0c105/ : FAILURE in 26m 52s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/ad612d6/ : SUCCESS in 42m 09s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/3de6d91/ : SUCCESS in 40m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d8890926091b283564e72a0e944bc09fb33996c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 16:20:09.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball http://logs.openstack.org/29/661229/4/check/build-openstack-puppet-tarball/94a64e2/ : SUCCESS in 1m 40s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-lint-ubuntu-bionic/3fcb1c4/ : SUCCESS in 3m 51s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-syntax-4-ubuntu-bionic/396639c/ : SUCCESS in 4m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-syntax-5-ubuntu-bionic/a04426d/ : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-syntax-6-ubuntu-bionic/804a4df/ : SUCCESS in 3m 49s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-4.8-centos-7/81a9524/ : SUCCESS in 28m 22s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-5.5-centos-7/098c5f0/ : SUCCESS in 25m 35s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-6.4-ubuntu-bionic/7321acb/ : SUCCESS in 11m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-latest-ubuntu-bionic/bcf8d16/ : SUCCESS in 12m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/cf3d74d/ : FAILURE in 34m 17s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/6d9583d/ : FAILURE in 29m 49s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-beaker-centos-7/433bb5f/ : SUCCESS in 23m 03s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-beaker-ubuntu-bionic/7663b29/ : FAILURE in 13m 24s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/29/661229/4/check/tripleo-puppet-ci-centos-7-undercloud-containers/353a66f/ : SUCCESS in 1h 37m 55s (non-voting)\n- puppet-swift-tripleo-standalone http://logs.openstack.org/29/661229/4/check/puppet-swift-tripleo-standalone/a6e3658/ : SUCCESS in 1h 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"328b098371f46ed7d53ef93e3c9f2846c267dec7","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-24 16:31:59.000000000","message":"Patch Set 4:\n\nError looks unrelated:\n  http://logs.openstack.org/29/661229/4/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/6d9583d/job-output.txt.gz#_2019-05-24_15_11_48_185507","accounts_in_message":[],"_revision_number":4},{"id":"5f4e686141644a2487b1c10de11c707e86080f71","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-24 18:25:41.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a55c94d54faecd1198b201ac5a45ad0364484351","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-24 19:16:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/ff745b4/ : SUCCESS in 33m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/38e7066/ : FAILURE in 22m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/b50e3c4/ : FAILURE in 48m 09s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/29/661229/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/99ec91d/ : SUCCESS in 29m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"18c7d51e05d20c9d7de8e5fae6e3e1b3844a4fa8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:03:45.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball http://logs.openstack.org/29/661229/4/check/build-openstack-puppet-tarball/b26e6d6/ : SUCCESS in 1m 43s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-lint-ubuntu-bionic/08c8c61/ : SUCCESS in 4m 13s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-syntax-4-ubuntu-bionic/e22866b/ : SUCCESS in 4m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-syntax-5-ubuntu-bionic/6260eee/ : SUCCESS in 4m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-syntax-6-ubuntu-bionic/1179d1e/ : SUCCESS in 3m 59s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-4.8-centos-7/22b4710/ : SUCCESS in 21m 17s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-5.5-centos-7/2db549c/ : SUCCESS in 19m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-6.4-ubuntu-bionic/2fdcce4/ : SUCCESS in 13m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-unit-latest-ubuntu-bionic/d99704b/ : SUCCESS in 11m 38s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/11f2c1c/ : FAILURE in 32m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/94bc80f/ : FAILURE in 32m 20s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/29/661229/4/check/puppet-openstack-beaker-centos-7/459fd33/ : SUCCESS in 24m 09s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/29/661229/4/check/puppet-openstack-beaker-ubuntu-bionic/9a0ea53/ : FAILURE in 18m 55s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/29/661229/4/check/tripleo-puppet-ci-centos-7-undercloud-containers/4e4edaf/ : SUCCESS in 1h 30m 38s (non-voting)\n- puppet-swift-tripleo-standalone http://logs.openstack.org/29/661229/4/check/puppet-swift-tripleo-standalone/3a14433/ : SUCCESS in 1h 07m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4c1afc548b4eee361b21f3c49c20718e6be8ac89","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-24 20:47:24.000000000","message":"Patch Set 4:\n\nhrm looks like scenario002 is broken for some reason, will need to look into this","accounts_in_message":[],"_revision_number":4},{"id":"82db1af2bb68de6e2df867f9875e0f306feff4c3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-24 21:25:11.000000000","message":"Patch Set 4: -Code-Review\n\nmight be related tho, swift is is erroring http://logs.openstack.org/29/661229/4/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/94bc80f/logs/glance/api.txt.gz#_2019-05-24_19_01_55_592","accounts_in_message":[],"_revision_number":4},{"id":"b5b6ec22196e7d81ecae8724232751a0d84651d4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-24 21:26:01.000000000","message":"Patch Set 4:\n\nMay 24 18:55:41 centos-7-rax-ord-0006548196 swift-proxy-server: ValueError: IPv6 addresses must be between \u0027[]\u0027\n\nhttp://logs.openstack.org/29/661229/4/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/94bc80f/logs/swift/swift.txt.gz#_May_24_18_55_41","accounts_in_message":[],"_revision_number":4},{"id":"577e0f61b6f04ee769ab5d2c3c89604bc64bff6f","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-05-27 11:31:42.000000000","message":"Abandoned\n\nI think for swift this is\u0027nt necessary. Ref: https://bugs.launchpad.net/swift/+bug/1610064","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6d1ab474acfa413d47fba186d5a415f557e66768","revisions":{"9ebbb999ef559b370a9668b40034de4c8259ce62":{"kind":"REWORK","_number":1,"created":"2019-05-24 08:04:35.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/29/661229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/29/661229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/1"}}},"commit":{"parents":[{"commit":"1fde777783a737af4c1bc67d035a5d514da35961","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1fde777783a737af4c1bc67d035a5d514da35961"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 08:03:23.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 08:03:23.000000000","tz":120},"subject":"Prefix memcached_servers IPv6 address with `inet6:`","message":"Prefix memcached_servers IPv6 address with `inet6:`\n\nDepends-On: https://review.opendev.org/661136\nChange-Id: I6e11480a8d122a7b5ae9325357e558a0416f676c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9ebbb999ef559b370a9668b40034de4c8259ce62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9ebbb999ef559b370a9668b40034de4c8259ce62"}]},"branch":"refs/heads/master"},"0c321ef8acb99ab97d98de4507299a1146862791":{"kind":"REWORK","_number":2,"created":"2019-05-24 08:13:36.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/29/661229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/29/661229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/2"}}},"commit":{"parents":[{"commit":"1fde777783a737af4c1bc67d035a5d514da35961","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1fde777783a737af4c1bc67d035a5d514da35961"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 08:03:23.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 08:13:24.000000000","tz":120},"subject":"Prefix memcached_servers IPv6 address with `inet6:`","message":"Prefix memcached_servers IPv6 address with `inet6:`\n\nDepends-On: https://review.opendev.org/661136\nChange-Id: I6e11480a8d122a7b5ae9325357e558a0416f676c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0c321ef8acb99ab97d98de4507299a1146862791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0c321ef8acb99ab97d98de4507299a1146862791"}]},"branch":"refs/heads/master"},"7dfe8a665f3fadf78bfc74b007e083b775c4c2e8":{"kind":"REWORK","_number":3,"created":"2019-05-24 09:39:29.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/29/661229/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/29/661229/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/3"}}},"commit":{"parents":[{"commit":"1fde777783a737af4c1bc67d035a5d514da35961","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1fde777783a737af4c1bc67d035a5d514da35961"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 08:03:23.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 09:39:17.000000000","tz":120},"subject":"Prefix memcached_servers IPv6 address with `inet6:`","message":"Prefix memcached_servers IPv6 address with `inet6:`\n\nDepends-On: https://review.opendev.org/661136\nChange-Id: I6e11480a8d122a7b5ae9325357e558a0416f676c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7dfe8a665f3fadf78bfc74b007e083b775c4c2e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7dfe8a665f3fadf78bfc74b007e083b775c4c2e8"}]},"branch":"refs/heads/master"},"6d1ab474acfa413d47fba186d5a415f557e66768":{"kind":"REWORK","_number":4,"created":"2019-05-24 11:59:58.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/29/661229/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/29/661229/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/29/661229/4"}}},"commit":{"parents":[{"commit":"1fde777783a737af4c1bc67d035a5d514da35961","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1fde777783a737af4c1bc67d035a5d514da35961"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 08:03:23.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-05-24 11:59:04.000000000","tz":120},"subject":"Prefix memcached_servers IPv6 address with `inet6:`","message":"Prefix memcached_servers IPv6 address with `inet6:`\n\nNew dependency: openstack/openstacklib\n\nDepends-On: https://review.opendev.org/661136\nChange-Id: I6e11480a8d122a7b5ae9325357e558a0416f676c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/6d1ab474acfa413d47fba186d5a415f557e66768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/6d1ab474acfa413d47fba186d5a415f557e66768"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
