)]}'
{"id":"openstack%2Fpuppet-swift~661419","triplet_id":"openstack%2Fpuppet-swift~master~I8a1bcdfbf4e9ea7c9d1fc26a10030dff80567f90","project":"openstack/puppet-swift","branch":"master","topic":"consistent-project-name","hashtags":[],"change_id":"I8a1bcdfbf4e9ea7c9d1fc26a10030dff80567f90","subject":"Consistent project name","status":"MERGED","created":"2019-05-24 23:24:47.000000000","updated":"2019-05-28 16:10:28.000000000","submitted":"2019-05-28 16:10:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661419-1559059828910-c3cf19bc","meta_rev_id":"73c0ee3f1c9f07c5ee81d6ba7f04a4c311b9f59d","_number":661419,"virtual_id_number":661419,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-28 16:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-25 00:17:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-05-28 14:34:30.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-05-28 01:14:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-05-28 14:34:30.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-25 00:17:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-28 01:14:03.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-05-28 14:34:30.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-05-28 16:10:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"19392076e77c6fdda264faba989eff78f4919595","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-05-24 23:24:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4462790243b0867030a8a4a6987f00062d5f3e1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-25 00:17:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/19/661419/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/e2a1f23/ : SUCCESS in 45m 20s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/19/661419/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/28a2228/ : SUCCESS in 45m 53s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/19/661419/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/232b4a9/ : FAILURE in 50m 11s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/19/661419/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/a88c842/ : FAILURE in 44m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef160a13a1ab4da74ca6955699a59d60153ecd2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 01:01:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/19/661419/1/check/puppet-openstack-lint-ubuntu-bionic/d3b8df8/ : SUCCESS in 6m 25s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/19/661419/1/check/puppet-openstack-syntax-4-ubuntu-bionic/3cc2873/ : SUCCESS in 6m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/19/661419/1/check/puppet-openstack-syntax-5-ubuntu-bionic/f0f4acf/ : SUCCESS in 6m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/19/661419/1/check/puppet-openstack-syntax-6-ubuntu-bionic/43d8054/ : SUCCESS in 6m 20s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/19/661419/1/check/puppet-openstack-unit-4.8-centos-7/09d555f/ : SUCCESS in 21m 55s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/19/661419/1/check/puppet-openstack-unit-5.5-centos-7/68f02fd/ : SUCCESS in 17m 26s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/19/661419/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/b746e09/ : SUCCESS in 15m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/19/661419/1/check/puppet-openstack-unit-latest-ubuntu-bionic/318e18b/ : SUCCESS in 14m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/19/661419/1/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/be1f2ce/ : FAILURE in 29m 34s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/19/661419/1/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/7537baa/ : SUCCESS in 1h 01m 32s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/19/661419/1/check/puppet-openstack-beaker-centos-7/9c79a7c/ : SUCCESS in 24m 55s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/19/661419/1/check/puppet-openstack-beaker-ubuntu-bionic/7bf051a/ : FAILURE in 16m 27s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/19/661419/1/check/tripleo-puppet-ci-centos-7-undercloud-containers/c08bfe4/ : SUCCESS in 1h 30m 24s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/19/661419/1/check/build-openstack-releasenotes/68d992a/html/ : SUCCESS in 2m 12s\n- puppet-swift-tripleo-standalone http://logs.openstack.org/19/661419/1/check/puppet-swift-tripleo-standalone/cfc2dea/ : SUCCESS in 1h 09m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc8bb9fd25db3c52421f94ed2173b40345878148","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-28 01:14:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24ce08a2fc3374517292aafbe10a74aaf933c897","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-05-28 14:34:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dc0a05cc9d656085a7617c5b54e4a4be0bb35e7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 14:34:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d96bf76af87bfa3933b6c75a58b79a153d209d31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 16:10:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/19/661419/1/gate/puppet-openstack-lint-ubuntu-bionic/d3337bd/ : SUCCESS in 3m 52s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/19/661419/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/3c20156/ : SUCCESS in 6m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/19/661419/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/177bb0b/ : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/19/661419/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/805c41c/ : SUCCESS in 6m 41s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/19/661419/1/gate/puppet-openstack-unit-4.8-centos-7/33a9f75/ : SUCCESS in 28m 09s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/19/661419/1/gate/puppet-openstack-unit-5.5-centos-7/c61dee2/ : SUCCESS in 25m 41s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/19/661419/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/ba6ab86/ : SUCCESS in 11m 56s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/19/661419/1/gate/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/d7b810e/ : FAILURE in 31m 47s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/19/661419/1/gate/puppet-openstack-integration-5-scenario002-tempest-centos-7/530de06/ : SUCCESS in 1h 03m 21s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/19/661419/1/gate/puppet-openstack-beaker-centos-7/157e48e/ : SUCCESS in 27m 05s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/19/661419/1/gate/puppet-openstack-beaker-ubuntu-bionic/46bec87/ : FAILURE in 17m 26s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/19/661419/1/gate/tripleo-puppet-ci-centos-7-undercloud-containers/e352520/ : SUCCESS in 1h 29m 36s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/19/661419/1/gate/build-openstack-releasenotes/3fc48c5/html/ : SUCCESS in 2m 38s\n- puppet-swift-tripleo-standalone http://logs.openstack.org/19/661419/1/gate/puppet-swift-tripleo-standalone/3949a5f/ : SUCCESS in 1h 09m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3f458d748d34c136f183be74b9cf6ae2771397a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 16:10:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1041c07684a9d105b0b166c3cda0d1083bc829ba","revisions":{"1041c07684a9d105b0b166c3cda0d1083bc829ba":{"kind":"REWORK","_number":1,"created":"2019-05-24 23:24:47.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/19/661419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/19/661419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/661419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/661419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/661419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/19/661419/1"}}},"commit":{"parents":[{"commit":"1fde777783a737af4c1bc67d035a5d514da35961","subject":"Fix client spec test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1fde777783a737af4c1bc67d035a5d514da35961"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-05-24 23:24:35.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-05-24 23:24:35.000000000","tz":120},"subject":"Consistent project name","message":"Consistent project name\n\nChange-Id: I8a1bcdfbf4e9ea7c9d1fc26a10030dff80567f90\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1041c07684a9d105b0b166c3cda0d1083bc829ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1041c07684a9d105b0b166c3cda0d1083bc829ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
