)]}'
{"id":"openstack%2Fpuppet-swift~665816","triplet_id":"openstack%2Fpuppet-swift~master~I1cf5dc4d8b0586d930116393f8a5c42b082feda8","project":"openstack/puppet-swift","branch":"master","topic":"665816","hashtags":[],"change_id":"I1cf5dc4d8b0586d930116393f8a5c42b082feda8","subject":"assert_private() params and deps classes","status":"ABANDONED","created":"2019-06-17 23:36:35.000000000","updated":"2020-09-18 22:57:36.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e8d4627fe75b38cf3b0ca8d215efab67e351016c","_number":665816,"virtual_id_number":665816,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-18 01:43:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-06-18 00:43:21.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-06-18 01:45:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 00:43:21.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 01:43:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-18 01:45:11.000000000","updated_by":{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},"reviewer":{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},"state":"REVIEWER"}],"messages":[{"id":"f71561b5647bcf00dcd18bd2d37d9c393bed3402","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:36:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5be83526646049506a4dc0da4347f194ee772254","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-06-17 23:53:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"813d78365634d7199d8a4255d0a7bb6970ee244d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 00:43:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/16/665816/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/8666327/ : SUCCESS in 36m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 puppet-openstack-integration-5-scenario002-tempest-fedora-28 : NODE_FAILURE in 0s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/16/665816/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/151f025/ : SUCCESS in 42m 53s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 puppet-openstack-integration-5-scenario004-tempest-fedora-28 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ff75897f7498c0ccb05d0f5d8cbf73570373e04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 01:43:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/16/665816/2/check/puppet-openstack-lint-ubuntu-bionic/6677a14/ : SUCCESS in 4m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/16/665816/2/check/puppet-openstack-syntax-5-ubuntu-bionic/7a1810d/ : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/16/665816/2/check/puppet-openstack-syntax-6-ubuntu-bionic/2c24cfe/ : SUCCESS in 4m 26s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/16/665816/2/check/puppet-openstack-unit-5.5-centos-7/8965d0b/ : FAILURE in 19m 08s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/16/665816/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/8702dbd/ : FAILURE in 11m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/16/665816/2/check/puppet-openstack-unit-latest-ubuntu-bionic/1d11577/ : FAILURE in 13m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic http://logs.openstack.org/16/665816/2/check/puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic/3406eed/ : FAILURE in 29m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 http://logs.openstack.org/16/665816/2/check/puppet-openstack-integration-5-scenario002-tempest-centos-7/564ff65/ : SUCCESS in 59m 20s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/16/665816/2/check/puppet-openstack-beaker-centos-7/5901b61/ : SUCCESS in 25m 08s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/16/665816/2/check/puppet-openstack-beaker-ubuntu-bionic/9e06580/ : FAILURE in 15m 07s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers http://logs.openstack.org/16/665816/2/check/tripleo-puppet-ci-centos-7-undercloud-containers/952ad72/ : SUCCESS in 1h 45m 12s (non-voting)\n- puppet-swift-tripleo-standalone http://logs.openstack.org/16/665816/2/check/puppet-swift-tripleo-standalone/75d757b/ : SUCCESS in 1h 05m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"41c8740ff44bcee88e5616165ec76002ba90eb1c","author":{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},"date":"2019-06-18 01:45:11.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89d3a0769fc3f8088e5aeca8e065d65009ffb25d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-18 22:57:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b4cf1c7644dbdd4d3a2443389d205fb097fc0976","revisions":{"321b64296c2c2ad1a9074c6d751d27fc844c1e40":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:36:35.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/665816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/16/665816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/665816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/665816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/665816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/16/665816/1"}}},"commit":{"parents":[{"commit":"9cec44f7cf5f925364809da15131fc5ed3ad6be6","subject":"Merge \"Deprecated swift::client_package_ensure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9cec44f7cf5f925364809da15131fc5ed3ad6be6"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:36:27.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:36:27.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I1cf5dc4d8b0586d930116393f8a5c42b082feda8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/321b64296c2c2ad1a9074c6d751d27fc844c1e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/321b64296c2c2ad1a9074c6d751d27fc844c1e40"}]},"branch":"refs/heads/master"},"b4cf1c7644dbdd4d3a2443389d205fb097fc0976":{"kind":"REWORK","_number":2,"created":"2019-06-17 23:53:40.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/665816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/16/665816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/665816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/665816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/665816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/16/665816/2"}}},"commit":{"parents":[{"commit":"9cec44f7cf5f925364809da15131fc5ed3ad6be6","subject":"Merge \"Deprecated swift::client_package_ensure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9cec44f7cf5f925364809da15131fc5ed3ad6be6"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:36:27.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-06-17 23:53:34.000000000","tz":120},"subject":"assert_private() params and deps classes","message":"assert_private() params and deps classes\n\nThese classes are private and should never be\nused outside of it\u0027s module.\n\nChange-Id: I1cf5dc4d8b0586d930116393f8a5c42b082feda8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b4cf1c7644dbdd4d3a2443389d205fb097fc0976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b4cf1c7644dbdd4d3a2443389d205fb097fc0976"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
