)]}'
{"id":"openstack%2Fpuppet-swift~672427","triplet_id":"openstack%2Fpuppet-swift~master~Ifbd46b76273c15208f74ce481c2f8adcfccb865f","project":"openstack/puppet-swift","branch":"master","topic":"prep-release-train-3","hashtags":[],"change_id":"Ifbd46b76273c15208f74ce481c2f8adcfccb865f","subject":"Prepare Train M3","status":"MERGED","created":"2019-07-24 08:20:25.000000000","updated":"2019-07-24 13:46:43.000000000","submitted":"2019-07-24 13:46:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672427-1563976003130-31ef7618","meta_rev_id":"a24db1941a40c11210c3f0877a689f68d7006994","_number":672427,"virtual_id_number":672427,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-24 13:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-07-24 13:32:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-07-24 13:32:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 13:32:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-24 13:46:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"597fb3ac1ccb0915a0c75490e84bba2822789e35","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-24 08:20:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70133ac29cd06e851d38f93110c68afd3e838b6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 08:49:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/27/672427/1/check/build-openstack-puppet-tarball/aa2c846/ : SUCCESS in 3m 03s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/27/672427/1/check/puppet-openstack-lint-ubuntu-bionic/e7bba93/ : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1553497bf30b3fd6e6c6cacd65650241abaa113e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-24 13:32:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0dfcffe73e3a3d2fd381865c36007053cb3e4f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:33:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"215471097b6d9f3546fe28e971729f76cd030476","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:46:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/27/672427/1/gate/build-openstack-puppet-tarball/e52b84a/ : SUCCESS in 2m 11s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/27/672427/1/gate/puppet-openstack-lint-ubuntu-bionic/e2a35b7/ : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"97dae3dbbfda142e45f42df9d7b040e3fc2466ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 13:46:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75bdc8189f5f2718671d72ce20f09d4023ebd6e7","revisions":{"75bdc8189f5f2718671d72ce20f09d4023ebd6e7":{"kind":"REWORK","_number":1,"created":"2019-07-24 08:20:25.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/27/672427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/27/672427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/672427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/672427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/672427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/27/672427/1"}}},"commit":{"parents":[{"commit":"9cec44f7cf5f925364809da15131fc5ed3ad6be6","subject":"Merge \"Deprecated swift::client_package_ensure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9cec44f7cf5f925364809da15131fc5ed3ad6be6"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-24 08:19:41.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-24 08:19:41.000000000","tz":480},"subject":"Prepare Train M3","message":"Prepare Train M3\n\nUpdate the version metadata for Train milestone 3\n\nChange-Id: Ifbd46b76273c15208f74ce481c2f8adcfccb865f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/75bdc8189f5f2718671d72ce20f09d4023ebd6e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/75bdc8189f5f2718671d72ce20f09d4023ebd6e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
