)]}'
{"id":"openstack%2Fpuppet-swift~676088","triplet_id":"openstack%2Fpuppet-swift~master~Ia51275bcae3c38969bf7764e4c9956f5f612230a","project":"openstack/puppet-swift","branch":"master","topic":"fix-openstackdocstheme","hashtags":[],"change_id":"Ia51275bcae3c38969bf7764e4c9956f5f612230a","subject":"Bump the openstackdocstheme extension to 1.20","status":"MERGED","created":"2019-08-13 03:30:15.000000000","updated":"2019-08-13 05:52:23.000000000","submitted":"2019-08-13 05:52:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676088-1565675543913-53e3b4f1","meta_rev_id":"b3da08c9b3b62fdcf757824c9076fef9ea3fff4c","_number":676088,"virtual_id_number":676088,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-08-13 05:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-08-13 05:43:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-08-13 05:43:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 05:52:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c255c891d9117050b333ae00c75d13fa6f7eb34f","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-13 03:30:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8599828162a06104044473107ed0f31742255c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 03:38:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/240c7d5017414489b332ef9dcc7fed7d : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a1975e60a5fcd7fc816798a2329ce6ceba08948a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-13 05:43:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b5d41f613010318a56b5fe7e43eef6da5ef3068","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 05:44:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"197d0cf129b68f1a9fab82e01aa07dd3802c6cc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 05:52:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e45c29eff79c8267a257ffce6b2f5a3f15597ff7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 05:52:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c3d869cdda14eecac8956c788c5027d : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ac50fe577dadb65c376d93912e9a8a4b43c97e3","revisions":{"5ac50fe577dadb65c376d93912e9a8a4b43c97e3":{"kind":"REWORK","_number":1,"created":"2019-08-13 03:30:15.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/88/676088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/88/676088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/88/676088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/88/676088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/88/676088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/88/676088/1"}}},"commit":{"parents":[{"commit":"573d1ca67ae8ecf60320da97c86343ea09eee65f","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/573d1ca67ae8ecf60320da97c86343ea09eee65f"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-13 03:29:08.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-13 03:29:08.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20[1]:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n\n[1]https://docs.openstack.org/openstackdocstheme/latest/#using-the-theme\n\nChange-Id: Ia51275bcae3c38969bf7764e4c9956f5f612230a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5ac50fe577dadb65c376d93912e9a8a4b43c97e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5ac50fe577dadb65c376d93912e9a8a4b43c97e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
