)]}'
{"id":"openstack%2Fpuppet-swift~679416","triplet_id":"openstack%2Fpuppet-swift~master~Ifa4e2ed894844ab1c7501f8edc18e55a9dea7e61","project":"openstack/puppet-swift","branch":"master","topic":"679416","hashtags":[],"change_id":"Ifa4e2ed894844ab1c7501f8edc18e55a9dea7e61","subject":"Add support for audit middleware","status":"MERGED","created":"2019-08-30 10:25:04.000000000","updated":"2020-02-06 18:39:13.000000000","submitted":"2020-02-06 18:37:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679416-1581014268529-b4850ae3","meta_rev_id":"42713c7480eb49cb5f5c98f71ee063bd2eaeb231","_number":679416,"virtual_id_number":679416,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"date":"2020-02-05 13:05:49.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2020-02-06 15:30:28.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 18:37:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-31 10:37:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-03 19:08:27.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-09-02 01:19:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2019-09-03 14:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2019-09-10 18:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-31 10:37:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-02 01:19:51.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2019-09-03 19:08:27.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-10 18:52:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-05 13:05:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-02-06 15:30:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-02-06 18:37:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab4c969ef41481da1838917ee8fb5dc4ae97c175","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2019-08-30 10:25:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83c74d0089e4786a78d52bee95bcd9d4d815f232","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-30 11:11:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/16/679416/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/71ca59d/ : SUCCESS in 39m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/16/679416/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/5de8e68/ : SUCCESS in 44m 08s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/16/679416/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/2220155/ : SUCCESS in 38m 57s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/16/679416/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/710b1e7/ : SUCCESS in 29m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee19cab08984c6514de82a331b8084a5033c4134","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 12:23:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0c4e35165da44b1883792c1e1c2ed9e : SUCCESS in 4m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da49f6c1f0f04f899524d79f3e554b63 : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb0aea2f6dc84d12840dcce17a7b9224 : SUCCESS in 5m 29s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7cbfa0f1b49a45d882e0306a8197bd53 : SUCCESS in 23m 04s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e91fe62e21444f9a7f00869fc86ee2e : SUCCESS in 17m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bdbffb4bd452469aa19c851410f96f32 : SUCCESS in 18m 44s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6954ca3b925e4102952eb9e15c680753 : FAILURE in 33m 18s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/59f2a2842cab44258beb4964c2787353 : SUCCESS in 1h 01m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1132efb37f6742dbbb089d883d431004 : SUCCESS in 27m 26s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c40a5b13b1e84def889d6fc88e802222 : FAILURE in 16m 46s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/03b1ca4481ce476baaf8d4cbbe904a87 : SUCCESS in 1h 54m 25s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2ea2b8053b1f4e72ae7d3d963f8dc59d : SUCCESS in 1h 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d7f9ccf5e896edd6905d4d49293f0adb3bcd9c4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-30 19:47:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b5344b16f037107f052eacb5c2023bc3e294a14","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-31 01:12:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"090eab97c4204553083b3ac82a646f21a3fc482a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-31 09:52:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"194a1601edd7e034ebbc9a9e9ea4253a0b5d4591","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-31 09:52:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2a23ea8b64f6abe7349efa2b2ed0aafcd6a0a395","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-31 09:53:28.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"aa3c8956c9a19ea77962c1491bf5c80566087973","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-08-31 09:54:17.000000000","message":"Patch Set 4:\n\nTook the liberty to clean it up a bit, add testing and release note, hope you don\u0027t mind :)","accounts_in_message":[],"_revision_number":4},{"id":"606bae1805113a88dc9d531baefe0494ee5cc76a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-31 10:37:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/16/679416/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/b415046/ : SUCCESS in 32m 55s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/16/679416/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/63c6f73/ : SUCCESS in 35m 14s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/16/679416/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/56ec269/ : SUCCESS in 42m 29s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/16/679416/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/fbb725f/ : SUCCESS in 32m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"73accfb51b9cd3dd39b38c092b99ddaf0abe4dfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 11:44:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc754cbe2f9d4f7fa9858ac2d42cdcb1 : SUCCESS in 5m 45s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33374280a98f44c885421feda8e0cc84 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8269c483e9a94a61940f13e7465ec9f3 : SUCCESS in 7m 50s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/baea3200623e487ab04d11415484f211 : SUCCESS in 18m 57s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2382b2c45baf4b9ba2438415d6e01a5a : SUCCESS in 16m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01a74f8185ba4c7e900d982ab2af5142 : SUCCESS in 19m 26s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c052859d2b0c4bce919f12c3a8f17a3d : FAILURE in 33m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/f19dac47b1f24b11a76a6707be9673da : SUCCESS in 1h 07m 22s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/93e2c83574944ef9b63ee7fe243ed89d : SUCCESS in 33m 10s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c4a437eff9e43b0a6332467b1bcbb5b : FAILURE in 19m 20s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d12c10e2f7014ba99c9535578e94e2bd : SUCCESS in 1h 49m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b64e51ba280849de8689339a38642ae5 : SUCCESS in 4m 09s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7455ee2511114ed2a4baf92550dd84c4 : SUCCESS in 1h 39m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99aa76c89ebab29a3c3abb556ae984408e6c9b26","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-02 01:19:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ed5033180ff251d272e9a1e1e879447cd8471e3d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-03 14:07:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c3c1197a6dc1d6dcf97588211ab9b90fd80b84c7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-03 19:08:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3494ef8305554718d2f022f5ef22bcd36b34472c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-09-10 18:52:18.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3c2440b9b02dfafd8ab7410ea6db7bbc0d9b2943","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-05 13:05:49.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c055f79491889e3e5a567ee6fb16d825f26cdd07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 14:56:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f75f1a71d1e64375898d8b3b52e7959a : SUCCESS in 5m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dad95bfd5a134a1a9b268baf73fd48ae : SUCCESS in 7m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b38857b50a14b2898ecfbdbbe692811 : SUCCESS in 5m 15s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b4c4ab43598a4b2ba8183229af2f5402 : SUCCESS in 18m 58s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc0518e37d1949e2b07d2534d5f3484c : SUCCESS in 17m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f72c893591a406898dcd4dd79eb5eb2 : SUCCESS in 15m 51s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21f7b20b3a9e483fa0231e50342ecce7 : FAILURE in 39m 24s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/13e1d5b79878470fa6a1919ec413aea3 : SUCCESS in 1h 08m 38s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cac4fbe77082424a919a5a966848f1e9 : FAILURE in 1h 06m 13s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/736d11e0d6e44ffab1b79827cd092b69 : SUCCESS in 30m 51s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/938e74abf8f4477bac3a32e9c00232f8 : FAILURE in 27m 09s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4f38827520674a789823975bb4f25df7 : SUCCESS in 1h 47m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a55596d90b68454f88c0a48ee87fa5dd : SUCCESS in 4m 59s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cbfce805a644429eaac42751046517f6 : SUCCESS in 1h 23m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8e13917ef6b10234c6b50a1f5b84c57e285131dd","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-06 15:30:28.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3941adc1b236952e85a4b6d7b18c571dfb2d20cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 17:01:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a4f29d2d90849d9a8ec0ff00027aeaa : SUCCESS in 5m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7554162043f34d8c8345f08dff34704c : SUCCESS in 6m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/078ed563c7e7400899f8e95edc199941 : SUCCESS in 7m 16s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/f5beb494076141238ff7100b303bb77c : SUCCESS in 22m 42s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/140eb9bb000243f883d5e1797f9c728a : SUCCESS in 18m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77862e5e74fc42cd91ffc0c00ac4c071 : SUCCESS in 19m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2789289a91e42049832f8eec059f6fd : FAILURE in 43m 24s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/98acfaa13b2a4d0e8a39ace6ed677c9e : SUCCESS in 1h 10m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/403e435f67ed4cd4bc556e97793e7859 : SUCCESS in 1h 04m 06s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a51176d371054042b56fc0faa39d6cb6 : SUCCESS in 30m 51s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b494c3b2154b477d9f8eed0524ce5878 : FAILURE in 28m 50s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f988fe58b6f44cd6937f6cbefb3d9e08 : FAILURE in 1h 24m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f467a2bc0ae4405199072350e847c185 : SUCCESS in 5m 35s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0e1d3b600efb4d469c67eb93f3dd3300 : FAILURE in 1h 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"21d914f3a28759c172d8dcdcab7bdec5186bed6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 17:01:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"60f23df477727774daac0198a92c56e2f1f9fc58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 18:37:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5830557a5697e42022fe608cb69e9dc13251a818","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 18:37:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/950d008a3f6843d098a117a2ec9112e1 : SUCCESS in 6m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c21d078b5cf4845b80c475d0beb175e : SUCCESS in 6m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fa419abf0dc4df8b7e1e7aab3674d76 : SUCCESS in 6m 20s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6c0e2d4f505945eb92f4c9dbfcda30e3 : SUCCESS in 24m 27s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a53fee775ba43b68a49a7990f439056 : SUCCESS in 16m 26s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2b59b83da134b85969fa0b917a93d3e : FAILURE in 44m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0f0001dd8b414638a46226ace64492ac : SUCCESS in 1h 08m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b4568b87eb954e958dd6815207819da8 : SUCCESS in 30m 01s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/1234664e6bc74695a6c6e49426daa26a : FAILURE in 1h 31m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46acfa75925943b1816e3f6cee611611 : SUCCESS in 4m 49s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/434b067e39974b5ca5b30b7ea7917035 : FAILURE in 1h 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9ad898ad33e004bb08120b87cbaf2581d1916c66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 18:39:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afd932dc1f274c71bab65fa8b421f25b : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2976c7d889cf7bc29a61e4bbf3fba662f08c3840","revisions":{"462010d0ac63ec5bf9d6051f779f5065dc7fe258":{"kind":"REWORK","_number":1,"created":"2019-08-30 10:25:04.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/16/679416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/16/679416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/1"}}},"commit":{"parents":[{"commit":"5ac50fe577dadb65c376d93912e9a8a4b43c97e3","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5ac50fe577dadb65c376d93912e9a8a4b43c97e3"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2019-08-30 10:24:22.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2019-08-30 10:24:22.000000000","tz":120},"subject":"Add support for audit middleware","message":"Add support for audit middleware\n\nDepends-On: Id1b2944783ecaaa631ce9c1ceebfa1a30de4d4c6\nChange-Id: Ifa4e2ed894844ab1c7501f8edc18e55a9dea7e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/462010d0ac63ec5bf9d6051f779f5065dc7fe258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/462010d0ac63ec5bf9d6051f779f5065dc7fe258"}]},"branch":"refs/heads/master"},"9876fd6fcc093a83baca06ce9d1b55c2e1cefa90":{"kind":"REWORK","_number":2,"created":"2019-08-31 09:52:16.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/679416/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/16/679416/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/2"}}},"commit":{"parents":[{"commit":"5ac50fe577dadb65c376d93912e9a8a4b43c97e3","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5ac50fe577dadb65c376d93912e9a8a4b43c97e3"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2019-08-30 10:24:22.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-08-31 09:51:52.000000000","tz":120},"subject":"Add support for audit middleware","message":"Add support for audit middleware\n\nCo-Authored-By: Tobias Urdin \u003ctobias.urdin@binero.se\u003e\nDepends-On: Id1b2944783ecaaa631ce9c1ceebfa1a30de4d4c6\nChange-Id: Ifa4e2ed894844ab1c7501f8edc18e55a9dea7e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9876fd6fcc093a83baca06ce9d1b55c2e1cefa90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9876fd6fcc093a83baca06ce9d1b55c2e1cefa90"}]},"branch":"refs/heads/master"},"572e7191df9e3918210b27cd96e55f5e91ab5f71":{"kind":"REWORK","_number":3,"created":"2019-08-31 09:52:59.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/679416/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/16/679416/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/3"}}},"commit":{"parents":[{"commit":"5ac50fe577dadb65c376d93912e9a8a4b43c97e3","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5ac50fe577dadb65c376d93912e9a8a4b43c97e3"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2019-08-30 10:24:22.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-08-31 09:52:53.000000000","tz":120},"subject":"Add support for audit middleware","message":"Add support for audit middleware\n\nCo-Authored-By: Tobias Urdin \u003ctobias.urdin@binero.se\u003e\nDepends-On: Id1b2944783ecaaa631ce9c1ceebfa1a30de4d4c6\nChange-Id: Ifa4e2ed894844ab1c7501f8edc18e55a9dea7e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/572e7191df9e3918210b27cd96e55f5e91ab5f71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/572e7191df9e3918210b27cd96e55f5e91ab5f71"}]},"branch":"refs/heads/master"},"2976c7d889cf7bc29a61e4bbf3fba662f08c3840":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-08-31 09:53:28.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/16/679416/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/16/679416/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/16/679416/4"}}},"commit":{"parents":[{"commit":"5ac50fe577dadb65c376d93912e9a8a4b43c97e3","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5ac50fe577dadb65c376d93912e9a8a4b43c97e3"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2019-08-30 10:24:22.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2019-08-31 09:53:15.000000000","tz":120},"subject":"Add support for audit middleware","message":"Add support for audit middleware\n\nCo-Authored-By: Christian Schwede \u003ccschwede@redhat.com\u003e\nDepends-On: Id1b2944783ecaaa631ce9c1ceebfa1a30de4d4c6\nChange-Id: Ifa4e2ed894844ab1c7501f8edc18e55a9dea7e61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2976c7d889cf7bc29a61e4bbf3fba662f08c3840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2976c7d889cf7bc29a61e4bbf3fba662f08c3840"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
