)]}'
{"id":"openstack%2Fpuppet-swift~693511","triplet_id":"openstack%2Fpuppet-swift~master~Ifbedd1b48f45f2d8970d4f6a787d0da4e442d693","project":"openstack/puppet-swift","branch":"master","topic":"memcache_max_connections","hashtags":[],"change_id":"Ifbedd1b48f45f2d8970d4f6a787d0da4e442d693","subject":"Add parameter memcache_max_connections","status":"MERGED","created":"2019-11-08 10:02:52.000000000","updated":"2019-11-15 14:53:01.000000000","submitted":"2019-11-15 14:51:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693511-1573829505001-bdb6f180","meta_rev_id":"554b847b6bd2969596a4de10442bedd38d30c7c2","_number":693511,"virtual_id_number":693511,"owner":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-15 14:51:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-13 18:08:14.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-11-15 13:00:52.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":2,"date":"2019-11-13 22:24:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-11-14 02:04:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-11-14 07:31:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-11-15 13:00:52.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-13 18:08:14.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-13 22:24:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-11-14 02:04:25.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-14 07:31:53.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2019-11-15 13:00:52.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-11-15 14:51:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f233798172a542c11b389c4bda49d34ec73a752f","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2019-11-08 10:02:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7440ad54576303506c4f43b64017209f4fbad4d0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-08 10:54:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/11/693511/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/c6b3ca5/ : SUCCESS in 36m 34s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/11/693511/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/8516103/ : SUCCESS in 46m 43s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/11/693511/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/c8fa831/ : SUCCESS in 49m 38s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/11/693511/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/82e0fe1/ : SUCCESS in 29m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a17a11add41a4b0af68665702506f8816546867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 11:52:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff3fc2c591134b8ba78d773633db7b7a : SUCCESS in 4m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eeedeb63905147018d5f769928957b9d : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a91e242649bf42c59c8d574054192491 : SUCCESS in 6m 42s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/25e87439b9ea4172b3c887c717208d6f : FAILURE in 12m 57s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63a9df2bd96c4b05bdbe3c1247576898 : FAILURE in 9m 42s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecc9428f9cd443f68a8e562839995ea2 : FAILURE in 10m 32s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea3980b11cd548e1a12cfc4de8b24145 : SUCCESS in 53m 22s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a08d587f9d834f83bdfb33ed45b762aa : SUCCESS in 1h 03m 40s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9bfcaebbbbb84bcca0751bfc806f8565 : SUCCESS in 25m 31s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6ce46387ca04a2bbd3d548e0515e040 : FAILURE in 28m 51s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/60c6c48dfdcd421999a20d1a0f5e98b9 : SUCCESS in 1h 44m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffec37a84d864e5eb072f041cf5a6952 : SUCCESS in 2m 50s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8526ede18c9d4f15961a717b58a63693 : SUCCESS in 1h 27m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad98c0b9fd4d3f395547b4870958272e2a6db24a","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2019-11-08 14:51:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"eea91ed53b4ce884545ca693bb449614d85e96bf","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2019-11-08 16:38:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"99c03cedbedc6ef5dfafc411d3eacf0e2ecf4bcc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-08 17:25:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/11/693511/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/07b50f9/ : SUCCESS in 38m 32s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/11/693511/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/423c370/ : SUCCESS in 45m 12s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/11/693511/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/9bbca90/ : SUCCESS in 37m 00s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/11/693511/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/9a6322d/ : SUCCESS in 31m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7821be86690e8f512796e12e627edc2703c216c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 18:26:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b07a42ab5296488dbc0b3530071ef688 : SUCCESS in 5m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ba0d1ff15c944278b20d5ec4f0eef53 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f0361aa4c9143a4a67fbeffd24d0170 : SUCCESS in 4m 22s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fd026a84c240451a9df170d7f1a049d6 : FAILURE in 14m 34s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef89fd24cc1e4a2d9947b611ca809594 : FAILURE in 12m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63f66885806d4c2da249ffdc248bb5a4 : FAILURE in 12m 08s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21cf2fe267e741189e0f6d22e4efa7ef : SUCCESS in 48m 11s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/26b0b48decbb44e4810bcdc7b2a1a664 : SUCCESS in 59m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1b53dcd977fe4b9e8ffa55488c98ab14 : SUCCESS in 24m 53s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9222def4acdb432fba6e50abc41a50fa : FAILURE in 25m 09s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6b1a4f6e6725426ea2f00030eaa25a47 : SUCCESS in 1h 43m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dba2cdac815442aa270fccc0a588004 : SUCCESS in 4m 21s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c92a8b533dd245cfbac268c014a5f47b : SUCCESS in 1h 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44afba6fccc636263b93ee9838507a551ef4cfda","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-11-08 20:48:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nThere\u0027s a syntax error (missing closing parenthesis).","accounts_in_message":[],"_revision_number":1},{"id":"671fc45e0ff840a008364ae54056f78c659b2d19","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2019-11-12 13:34:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3bf858ed08680a0a2f16ac6a28188a2be59ee83f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-12 14:22:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/11/693511/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/d052cb5/ : SUCCESS in 43m 52s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/11/693511/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/0144ded/ : SUCCESS in 45m 52s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/11/693511/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/d46e786/ : SUCCESS in 45m 50s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/11/693511/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/66fc627/ : SUCCESS in 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d9f5980e34a8f186405ccd4585c82f246672587a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 15:22:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a694f440cb2b453aa1b6c55bfd064a6c : SUCCESS in 4m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f6fa8233b754c098c42d8273b17180c : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d3e25884c01453ab38a21ba79a42eae : SUCCESS in 4m 47s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/20891f45d6f0407795b829cb42fcaecf : FAILURE in 24m 44s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7a1d19362bf437db190557dc0e35cd4 : FAILURE in 12m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b082d42d09af4307acbc7d78bec46884 : FAILURE in 17m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86835849732341d89be14178cf79a307 : SUCCESS in 45m 49s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/054cd455273e49f5a7ac45f72021873b : SUCCESS in 59m 22s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3dc113c1eef64f6a9cd86262272ae245 : SUCCESS in 26m 52s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b6517ea56504c5a8cb044002b2ff346 : FAILURE in 26m 22s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/36e1f2c17dee4660ba90fcd4a0d29381 : SUCCESS in 1h 45m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a46b5083aec94bb3a83009dfbb1219d7 : SUCCESS in 3m 46s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b4ffe55de71347d9b236df71be624359 : SUCCESS in 1h 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dce2de3b6701ab6c34998c8cf02ee65f4614e671","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2019-11-13 07:52:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1361be4b6bdfee9894da731ecbaae207b23c572d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-13 08:40:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/11/693511/3/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/a618227/ : SUCCESS in 40m 04s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/11/693511/3/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/b0f7c54/ : SUCCESS in 45m 12s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/11/693511/3/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/f2da324/ : SUCCESS in 46m 49s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/11/693511/3/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/5294cdd/ : FAILURE in 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2656b3492a577e9d36398356975a11afaaec4207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 09:26:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35bd379c211040529700514164d07d06 : SUCCESS in 6m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09e190f212274d09bf48ad7c3f7685f5 : SUCCESS in 6m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7f603ce8f3a41fb97564f690e58787d : SUCCESS in 6m 09s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/21b6f9d881d64df98f0fed9372eeb176 : FAILURE in 19m 49s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40e7ae89ba8048fd9f25648b1ca48547 : FAILURE in 13m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/462fbb7074b54666b6868bf1f3042f28 : FAILURE in 14m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d0a0d2cbde44c2bb2d0f5825446e4e3 : SUCCESS in 52m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/86d18e8e795a4007aa606a53b6671c1c : SUCCESS in 1h 03m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/679fd22709284d4aa47d8877c9e6d312 : SUCCESS in 26m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a191ccc34374585ba914e4f4a6ec3a5 : FAILURE in 25m 04s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8d722f8b7ff74069ac11361093d0fe8d : SUCCESS in 1h 28m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2208b07b9c7a4682838843a40aaf154c : SUCCESS in 3m 11s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4de36157e8074873a07d241435af8a0e : SUCCESS in 1h 32m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"80687454d8c391b1d21019271dbf5db5aaeadc4f","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2019-11-13 17:12:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"99af7040388e2af647e7b733eba600ea380babc3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-13 18:08:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/11/693511/4/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/e837a73/ : SUCCESS in 41m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/11/693511/4/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/8fc6be4/ : SUCCESS in 54m 25s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/11/693511/4/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/ff19f22/ : SUCCESS in 50m 01s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/11/693511/4/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/4a2ef5a/ : SUCCESS in 36m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"419fecc485bea857d26a01e4f98b60c4248fe380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 19:08:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce54afc721df4dcc8d1745956b66d752 : SUCCESS in 5m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b31eae1fa3843b89e52cb5cffacc43c : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5288ea52a0f74acaa62f0f8703af4166 : SUCCESS in 4m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/30beff43159d497ab8866f7947d1cda2 : SUCCESS in 24m 41s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9e2c81470e5439a8910ac7235d24aec : SUCCESS in 16m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/218f4b9a87a1431787accff545b59b54 : SUCCESS in 17m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/414334dd62e54309b9f43bd883d6e6fe : SUCCESS in 47m 47s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9d33ca39dff84c158a2a99c078f35673 : SUCCESS in 59m 39s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/40b72fb48f944d2496c113b5c3f8cdb1 : SUCCESS in 26m 34s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13880ec1912d432d93b86931cad2faaa : FAILURE in 31m 49s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4ac9670097c34b61b1045dbf6cfac9d1 : SUCCESS in 1h 46m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/997eed1c861448c49809c9807d4f456a : SUCCESS in 3m 50s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0d1763e06bf5463384ff11ccaf391af8 : FAILURE in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7addbafe720b4b0d683e4f23ca9a8c86e2492752","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-11-13 22:24:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f968e22d020189f78d755bbb263aa91b5d5a5548","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-14 02:04:25.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"56a4bd89d845fc3a730e609997b849c763cc929a","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2019-11-14 07:31:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2eb14fb1b7abde7b1ca52cca9ea77be110094d07","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-11-15 13:00:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9a330478c7f106e10cc7738b03890ce77e79f1a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 13:01:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e214a5c14ff10615f408889c148fc674f8e2eb21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 14:51:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d7c7667e5964ce0accc03aa9d84719f : SUCCESS in 4m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0f8a9c23851446883922929ae75c876 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a64463c9d984c66a8af4ac397fc3ee5 : SUCCESS in 11m 51s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4dd92ecf01864bf9938ef384afe57d8a : SUCCESS in 19m 36s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/739d972f89db40abbb6677ce38d7b2d5 : SUCCESS in 12m 36s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b17b512fc144b06b2dac447b088a0fc : SUCCESS in 54m 58s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e8ef4d6d72604386897dede0f9319c30 : SUCCESS in 59m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6faf79b63d1242b1a8c0688bbbd440ab : SUCCESS in 26m 31s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/dccdfab27c334eb4942a2a526404aec6 : FAILURE in 1h 45m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ecf328a3e91492b8ed01ed5188bfc1c : SUCCESS in 4m 18s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/543b2a5fe9494969930a4305a4017973 : SUCCESS in 1h 27m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4da19b80dfb521f2773bffce7297778ef0ec28f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 14:51:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3f156bf4cbc5c2a9d6c360ca9623fb9f0d7bb8bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 14:53:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d551adca48f49ff945247a2d3e13e9b : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2a6175e2f7b1300ef26e0c12172a53a81cdeb166","revisions":{"518b2125eeb045fc9225b7821fb5304a96cf2334":{"kind":"REWORK","_number":1,"created":"2019-11-08 10:02:52.000000000","uploader":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"ref":"refs/changes/11/693511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/11/693511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/1"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","date":"2019-11-08 09:58:54.000000000","tz":60},"committer":{"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","date":"2019-11-08 09:58:54.000000000","tz":60},"subject":"Add parameter memcache_max_connections","message":"Add parameter memcache_max_connections\n\nThis add parameter memcache_max_connections as new parameter for class\nproxy::cache.\n\nChange-Id: Ifbedd1b48f45f2d8970d4f6a787d0da4e442d693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/518b2125eeb045fc9225b7821fb5304a96cf2334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/518b2125eeb045fc9225b7821fb5304a96cf2334"}]},"branch":"refs/heads/master"},"e37b7c7fec2860f1c633d9abb19096a69c8002c5":{"kind":"REWORK","_number":2,"created":"2019-11-12 13:34:49.000000000","uploader":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"ref":"refs/changes/11/693511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/11/693511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/2"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","date":"2019-11-08 09:58:54.000000000","tz":60},"committer":{"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","date":"2019-11-12 13:33:42.000000000","tz":60},"subject":"Add parameter memcache_max_connections","message":"Add parameter memcache_max_connections\n\nThis add parameter memcache_max_connections as new parameter for class\nproxy::cache.\n\nChange-Id: Ifbedd1b48f45f2d8970d4f6a787d0da4e442d693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e37b7c7fec2860f1c633d9abb19096a69c8002c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e37b7c7fec2860f1c633d9abb19096a69c8002c5"}]},"branch":"refs/heads/master"},"007e95ad743018fb5299a32b2ace859c339de100":{"kind":"REWORK","_number":3,"created":"2019-11-13 07:52:18.000000000","uploader":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"ref":"refs/changes/11/693511/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/11/693511/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/3"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","date":"2019-11-08 09:58:54.000000000","tz":60},"committer":{"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","date":"2019-11-13 07:51:44.000000000","tz":60},"subject":"Add parameter memcache_max_connections","message":"Add parameter memcache_max_connections\n\nThis add parameter memcache_max_connections as new parameter for class\nproxy::cache.\n\nChange-Id: Ifbedd1b48f45f2d8970d4f6a787d0da4e442d693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/007e95ad743018fb5299a32b2ace859c339de100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/007e95ad743018fb5299a32b2ace859c339de100"}]},"branch":"refs/heads/master"},"2a6175e2f7b1300ef26e0c12172a53a81cdeb166":{"kind":"REWORK","_number":4,"created":"2019-11-13 17:12:20.000000000","uploader":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"ref":"refs/changes/11/693511/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/11/693511/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/11/693511/4"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","date":"2019-11-08 09:58:54.000000000","tz":60},"committer":{"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","date":"2019-11-13 17:11:55.000000000","tz":60},"subject":"Add parameter memcache_max_connections","message":"Add parameter memcache_max_connections\n\nThis add parameter memcache_max_connections as new parameter for class\nproxy::cache.\n\nChange-Id: Ifbedd1b48f45f2d8970d4f6a787d0da4e442d693\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2a6175e2f7b1300ef26e0c12172a53a81cdeb166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2a6175e2f7b1300ef26e0c12172a53a81cdeb166"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
