)]}'
{"id":"openstack%2Fpuppet-swift~694451","triplet_id":"openstack%2Fpuppet-swift~master~Iae3705db38ebaddf92ee6e80c20c1011f8f66706","project":"openstack/puppet-swift","branch":"master","topic":"recon-cron","hashtags":[],"change_id":"Iae3705db38ebaddf92ee6e80c20c1011f8f66706","subject":"Add support for swift-recon-cron job","status":"MERGED","created":"2019-11-15 05:00:02.000000000","updated":"2020-02-07 06:08:35.000000000","submitted":"2020-02-07 06:05:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694451-1581055546330-b4850ae3","meta_rev_id":"49a053a51aaa4f3ee407064690b8de1f5956e665","_number":694451,"virtual_id_number":694451,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-02-07 02:28:05.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 06:05:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-05 13:40:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-02-03 18:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-02-05 13:40:21.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 08:47:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-03 18:38:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-05 13:40:21.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-07 06:05:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"724f5b0ab6f7d293b17f29f51c55bb481d782039","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-11-15 05:00:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d450c95362fc5979361442f4745d960b60cecb2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-15 05:54:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/51/694451/1/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/382148d/ : SUCCESS in 42m 46s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/51/694451/1/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/41fcf31/ : SUCCESS in 52m 28s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/51/694451/1/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/e0b0c35/ : SUCCESS in 50m 27s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/51/694451/1/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/5b63965/ : SUCCESS in 37m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a371e881db6b7a03ca58971a25cf6c7a39d2c11d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 06:56:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/236dc623aa394d4ca492532d5da0e110 : FAILURE in 5m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77cbd1113c1b4d9cb50dc7340d497172 : FAILURE in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/757e7c447d92439bac07d148c2577326 : FAILURE in 4m 28s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/fb966a9492f04122aacfe748644d2061 : FAILURE in 10m 47s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e85fccbc2446468e9445798c13001dcb : FAILURE in 10m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/751814ca690d4cfd9810bc55fdb0dd67 : FAILURE in 8m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42592046c9a14dea8f9672f645b2c139 : SUCCESS in 42m 47s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2d884314297d4ac6bce3ab33ad1b0023 : SUCCESS in 1h 19m 07s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3b9ac6845bf5469db7700689b4d7c6f9 : SUCCESS in 23m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f9319ff3fd8453d8982a28957ee654d : FAILURE in 24m 40s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3cf234eea2aa4488a8fca44e317d3359 : FAILURE in 1h 50m 42s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d57da7b3b91d4998b7287ed3e75fe8e2 : SUCCESS in 1h 23m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4d5187a3d71c5acce806be8aa4c8a0196b7d1fb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-11-15 07:04:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bbb2fac8c5236e32965105a4be7e9afb0d3b6425","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-15 07:54:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/51/694451/2/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/75fcf81/ : SUCCESS in 44m 59s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/51/694451/2/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/31c9c16/ : SUCCESS in 41m 36s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/51/694451/2/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/1d827d4/ : FAILURE in 48m 01s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/51/694451/2/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/aebe06e/ : SUCCESS in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"35d5e81cc4b907600ae17b2197f74d30ef6ecc77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 08:49:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34ba8770c20a4f3280260fa7d3ef55dd : SUCCESS in 7m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2276a15bc64c43ae80a34e8dd9ae6103 : FAILURE in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb1331dadfce422fbe573298c8c0603b : FAILURE in 5m 04s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d25616061a8e44b8b4f95e4cd69a094b : FAILURE in 11m 32s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/126591f2a09546159b43d225dca45b43 : FAILURE in 11m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7698c869aa0344aaafc16938f4d48a34 : FAILURE in 14m 26s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7526257962f046a5966714d9d5caa974 : SUCCESS in 56m 50s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4a9fb94c320843d58896e0a165ba349e : SUCCESS in 57m 16s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/30c6221d8f024533afb656f2040ca315 : SUCCESS in 28m 51s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/913913dee9f64cbca2bd342ad9db39ff : FAILURE in 27m 39s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c9ba21b40ffd46f7b2a92c0a1e6a1739 : FAILURE in 1h 38m 38s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f7684709c8604cdcbcee7b199ce830fb : SUCCESS in 1h 28m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4155474a61dc2af4424dd64fd7ad9d4c95d57065","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-11-18 02:28:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c2a5ec910ddea2fa057162997f22cf21fcb27913","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-11-18 02:30:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c9c16b8556771defc2aeba86d40b43bbfffc3275","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-11-18 02:55:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"17bc3fa88791257f477110c21f63d0695b746aad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-18 03:39:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/51/694451/5/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/626faaf/ : SUCCESS in 39m 12s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/51/694451/5/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/dd6829d/ : SUCCESS in 41m 11s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/51/694451/5/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/3621aa4/ : SUCCESS in 42m 03s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/51/694451/5/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/205ffed/ : SUCCESS in 34m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b6e073c394afa4345bce836e5936a92aa07dee03","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-11-18 04:11:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ac80ebb8b071c1a811d05291903f91013d9dc932","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-18 05:01:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/51/694451/6/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/cc51cbd/ : SUCCESS in 35m 52s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/51/694451/6/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/74a08f9/ : SUCCESS in 48m 07s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/51/694451/6/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/115ac44/ : SUCCESS in 49m 04s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/51/694451/6/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/d934b9a/ : SUCCESS in 36m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fb972d8ad0d836ec17c0e19c202d7735c807c1a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-11-18 05:15:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8b25daa880809f2c6e907be3447da47cdb0d9f55","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-11-18 05:56:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b55e07f1231c33fe28a475d1155c229df7d51c50","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-18 06:39:45.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/51/694451/8/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/8948019/ : SUCCESS in 41m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/51/694451/8/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/eb2df21/ : SUCCESS in 39m 54s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/51/694451/8/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/28e299c/ : SUCCESS in 39m 23s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/51/694451/8/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/d3dafbb/ : SUCCESS in 32m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"47e3e1e733e54eb102e2ee7faba2cb9bf4f7257b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 07:51:57.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/042877adcdb6437caf01caebd9e45de9 : SUCCESS in 6m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d71c831aa1314433be0e66dc0e405f54 : SUCCESS in 5m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be53ff2d21864524a13c1386f7a5418b : SUCCESS in 5m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a9db2215eec8480797322b177a6af14c : FAILURE in 11m 27s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e83e0cfd90e47a9b8025d1cd6c09aa8 : FAILURE in 10m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e554212eebc44441a2cefa6d7f99d10c : FAILURE in 10m 45s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04a211226bb447e2b0544db08d4b9e3f : SUCCESS in 56m 10s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/08d6ca0854514a6590fe55dc018b50c0 : SUCCESS in 59m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a9600e85cb1f46f2b1ded2b379cef9f4 : SUCCESS in 29m 36s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1524e2cf3f2b4354a93d8b3b47aa25b6 : FAILURE in 31m 43s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/93a5f37890f745d7a9b80f549449ab8f : SUCCESS in 1h 52m 10s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0d335b8bd39f4a92b30a3a2daf76d17f : SUCCESS in 1h 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0da422a6df1dd0b44acf4baca386aa245081b02b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2019-11-18 07:57:48.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"478c1456c1d2fb69d19a30f84829abfc16b76a1b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-18 08:47:03.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-integration-5-scenario001-tempest-fedora-28 http://logs.rdoproject.org/51/694451/9/openstack-check-rdo/puppet-openstack-integration-5-scenario001-tempest-fedora-28/5558977/ : SUCCESS in 37m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-fedora-28 http://logs.rdoproject.org/51/694451/9/openstack-check-rdo/puppet-openstack-integration-5-scenario002-tempest-fedora-28/e2a8e7c/ : SUCCESS in 47m 43s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-fedora-28 http://logs.rdoproject.org/51/694451/9/openstack-check-rdo/puppet-openstack-integration-5-scenario003-tempest-fedora-28/a654237/ : SUCCESS in 47m 05s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-fedora-28 http://logs.rdoproject.org/51/694451/9/openstack-check-rdo/puppet-openstack-integration-5-scenario004-tempest-fedora-28/399459b/ : SUCCESS in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"dcf731014765639ca217b4ea2ea8c264691bca0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 09:52:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55de5b9d5d0a4a02bc9a43ae987fa868 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd5ffc10ed2843dc910f1573a266b9ea : SUCCESS in 3m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f27adb28592341f1bf579484be7f3dd1 : SUCCESS in 3m 55s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/396e279059b6487b8cb01a478f2ea5a2 : FAILURE in 11m 02s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9a7efd7952b47ff9f8760ae149d8ef2 : FAILURE in 9m 13s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a15a8c5123347aba8800b8ba61be461 : FAILURE in 9m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ac6b26b2d4444e781a8284ca14196ee : SUCCESS in 41m 42s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cabd4ee61c1d4b4689302c5970afa0f7 : SUCCESS in 59m 07s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/95ec89d15c1642dbbc817c18f5562432 : SUCCESS in 24m 02s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b65ad6b621a4457999a972651316afb : FAILURE in 25m 12s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4d1d86ab28e64073a2d3b063ace61c80 : SUCCESS in 1h 51m 18s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9de78ef9fa484b129414ad8eca1e6a2c : SUCCESS in 1h 37m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ef81339a936aa02635d38d3bd3de8c9c3259655b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-01-30 13:39:15.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"20b2e84528e35f8c9f7586da143606b3de987497","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-01-30 13:48:15.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"03d7d199f00cddba0025b8facb32e44c31bc0dd4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-01-30 14:12:35.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"1016faa87443f690afb0a9620c3d2662d5d8ab93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 16:23:05.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f73306c8c3e2465b92b03eadcec905ea : SUCCESS in 6m 47s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d987536dbda144dca85e09da10b12f00 : SUCCESS in 4m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0eaa52b7d654fe39b49ee057d980e4b : SUCCESS in 5m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a83344e9bf2f4390bbc172a70d78a969 : FAILURE in 27m 31s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bbd7f058e2b4601ab9c727acb9cb82b : FAILURE in 18m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f284bcd7eea449e2b56c97dad699fb9f : FAILURE in 17m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95fcaadd216e4cc2bfba29719bf360b4 : SUCCESS in 55m 06s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c2380f8bdf5f4b13b9e753768e9336b0 : SUCCESS in 1h 06m 32s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4f89ec52391d4b87a70b37cdc1bd03cd : SUCCESS in 1h 04m 35s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ee90f5e9aa3445c6bb195f259126543f : SUCCESS in 33m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2b41021bac244568e37e58d9d135e3f : FAILURE in 24m 52s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f64fd7be65de4597823c95affde6d3b9 : SUCCESS in 2h 05m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aab3c8a4ea1846cca460da50d6158143 : SUCCESS in 4m 44s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fdbe94586328483db533a9be4dfa63d0 : SUCCESS in 1h 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8578c572ddcec61d8630eeb245bc8d5a5b37d50e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-01-30 23:37:38.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"78a2ec3d207249ced81acbf408c37bb35610f295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 01:52:14.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae410ee1bb6f4f38ae8483e1607c8f83 : SUCCESS in 5m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f04b31ced0f842e28ddfb8eb93260c7e : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98dba63c0a554e9aa6908cc0bc8f5f06 : SUCCESS in 4m 30s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b847148dcef840d38194a92eaca9ddb2 : FAILURE in 22m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dfd7963e61643c4bc076106497ddce4 : FAILURE in 18m 03s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/897e074e412d46079f936c3aae4ddf5f : FAILURE in 15m 36s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59d0723ed52840a0b1499b6e5f42ce7a : SUCCESS in 48m 04s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/93af7b4319f541e4bed28556d754d0d2 : SUCCESS in 1h 02m 29s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/90a801bce51d4a0db81ebf80820caa16 : SUCCESS in 1h 08m 41s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6683ad3c308642bbb65afc8f1038c48d : SUCCESS in 27m 47s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca238413df334d83aaf46dc06c693b57 : FAILURE in 34m 59s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/11f8136493804d299bf15e0254f52fa6 : SUCCESS in 2h 04m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9de87a7a9dcb48b8a1ba973bc18e80bc : SUCCESS in 3m 30s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d2461277347042d0993b7384d49a4aa8 : SUCCESS in 1h 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"9c764c9cda65ab55b115b0e8436699e88e0dd132","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-01-31 02:05:20.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"19fa8b5f57f4a08e880222ba0f9edea95d0e20d2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-01-31 03:14:19.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"ce074aae4a6fd635ff7a966652f82b3bd17f7b99","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-01-31 04:34:10.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"e3420f5130a6fe0c69126f5133b13af6d8716974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 06:23:24.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccf15155fb0b4be4a6d0416df0ce3e35 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e3e88b05cf9457aa2b36e57a3034d04 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3a24d9e26214602948b70ba410b5dc0 : SUCCESS in 5m 54s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/8b252df796d94403b51cd1b3ef989d0d : FAILURE in 25m 37s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b6d0ce2661a42ce9c07147e84bb2a85 : FAILURE in 16m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/55f9b215a47f403eafcd41c1e948bbea : FAILURE in 14m 32s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11b8765bd52242f083cd4a9e5d3bc3dc : SUCCESS in 46m 47s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/593f5217faf8408fbed9f9d43d16a4c1 : SUCCESS in 59m 52s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8cc3dac0728a460f990d68a9cb533557 : SUCCESS in 1h 07m 38s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/23d76b16cdf04579b2e9d8ffb645e1b7 : SUCCESS in 27m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12d96e4e936941cb8faa936d2daafb90 : FAILURE in 21m 57s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4bc1a96069c645dcadc7b38d24aff0eb : SUCCESS in 1h 44m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cbda636c6ee4aec9c044362d8243343 : SUCCESS in 3m 26s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/613f8bd32f5847b1b6bac1d847988e28 : SUCCESS in 1h 17m 46s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"59c66e13c95526cc2ed5fa3205efc05ea890b32f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-01-31 07:57:29.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"b88bde26e442c58406ed10b5af94f755c9a68307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 09:58:07.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b622bd5c9e14251b2dde6f1ed830233 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/675ecdf1783c4c938d1c2710f196f18d : SUCCESS in 4m 25s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cbec38236e24e79ab8f76c12b885014 : SUCCESS in 4m 07s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e84d18c0e8a543fb99d8c86dcb347073 : SUCCESS in 26m 25s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8c079158ba240fcbe1b4f829c5e57a7 : SUCCESS in 17m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2285170b21fd4a9c9b38c5b5f1b6c9f2 : SUCCESS in 15m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/890e8903173b448b8fe405b0e199926d : SUCCESS in 1h 25m 47s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/065111339f504c2b8032e397868a2ff8 : SUCCESS in 1h 05m 12s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e7d2f81e8a8d4f33bf71c09dd28c4e84 : SUCCESS in 1h 03m 31s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/17edc11edce04793abdbb16fddef3cfe : SUCCESS in 31m 35s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6c41de2313944399ecaeb12e47ce559 : FAILURE in 45m 08s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/567dbe680e64420181793206777fe55b : SUCCESS in 1h 55m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f72e1187ea284e49b17f6cac9505ae5b : SUCCESS in 4m 31s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/84b6aff4ef6342eabdac1248ecef664c : SUCCESS in 1h 21m 54s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"fc0c31dd66a4aac74ff77847f801856f9d62869e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-03 18:38:24.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"cf9e14f6c9fc21f0622a9e99bc442c69bc37e2ee","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-05 13:40:21.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"5b7f2fad06c7cf71f86a4ea08f1cc42fd2abab5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 13:40:33.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"e26d7142fa34314685499206551c6fd6835f9bf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 15:56:13.000000000","message":"Patch Set 17: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b75dc43d59f4451f96916053d10d5eec : SUCCESS in 6m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d95ea254102c49238c617cfda89d57c2 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a995d19e2f514360aa3404d67b0ffd37 : SUCCESS in 6m 16s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/5f4cf1bdcac348b68e5578c4b25f015c : SUCCESS in 24m 54s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5eca3604acbf4ea7b66126020c813473 : SUCCESS in 19m 14s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adda91e35daf499b9f1968ec805b6c9d : FAILURE in 40m 28s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0a67092a4a9d47ce97647af7c0566acb : SUCCESS in 1h 14m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b4ee612ea365424e8015aca3b8c3a1e5 : SUCCESS in 29m 12s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/356ea8df4ac940058fa753dd4c9a5b0d : SUCCESS in 2h 12m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ae7e7b94258419fa231eedac98068fb : SUCCESS in 3m 22s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f27f355d543b41e39fa87df1ff0625d5 : SUCCESS in 1h 20m 04s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"03036e66757ae7e480aa02fe9b200e80559a357b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-05 21:46:12.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"9bfd599459adf18d882a855d024316dad0e9e8b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 23:33:04.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e035d436e8946bf8e385137b750fd0a : SUCCESS in 7m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ab269101eb649de85b6ff46fc26afe9 : SUCCESS in 8m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e373698b7234b998b2de894cbf55427 : SUCCESS in 8m 02s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/dfed36a814304f8eaddb1db7b3bb44c9 : SUCCESS in 24m 15s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b0bb72a7e0a41a1bf31d0660751941c : SUCCESS in 17m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36b3b39f110e42918e02de043221d74a : SUCCESS in 19m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60c97c8e1b2c4e30a41d4373f4dd292a : FAILURE in 37m 14s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8278141b18344cdaba563705d17cce49 : SUCCESS in 1h 01m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ab017b1a537049508b060906029c8079 : SUCCESS in 1h 04m 04s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0ed09df62afe45628098e04b6a601d1f : SUCCESS in 32m 53s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd4e920030e245f5ad8e9f30fce4f8ca : FAILURE in 28m 20s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6cfb80abbcf842e68f8caeb358aa268d : SUCCESS in 1h 40m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5670e97f1c744118934a4138d5e127a : SUCCESS in 5m 41s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a2d735acce0b4b2e9d6e53b33e57b756 : SUCCESS in 1h 35m 19s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9b43c22d8331589debc901b9af3bde8847b6786a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-07 02:28:05.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"35b773c1893800626ee1726213ba9224c2cf5400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 04:20:25.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebdd380a69a54b2d85eda68f2a8ce9c0 : SUCCESS in 4m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d443b163777432b8c1e19539fa5dfe7 : SUCCESS in 8m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/996c55ab50c6418fa89c6dbec7f2b516 : SUCCESS in 4m 01s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/542b8f8c729e4c4591c8dd627065933a : SUCCESS in 24m 45s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c49afb7c5c9345d6b143aa92123769a2 : SUCCESS in 14m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0e1559d6c284e6289ffc40124052cbd : SUCCESS in 13m 46s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ec2d7fdeb814ff48817b12e67f95cec : FAILURE in 33m 01s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c11d30b5faa04abba82d66ff842d0002 : SUCCESS in 1h 07m 36s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/40dcca2ba18c4bfcb9d406922e515ba0 : SUCCESS in 1h 06m 41s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/27f2e298e70c457798d542f7bb05db13 : SUCCESS in 28m 14s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afe7a888e48b4643934c84d851919f35 : FAILURE in 22m 00s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d6db7acdcb7b4c8aad7a655e861a8aec : SUCCESS in 1h 47m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48f77872e7a24c0f99ba6a0367f56b6d : SUCCESS in 3m 32s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/96757ea4e40f44a8997dcd0e73ef8278 : SUCCESS in 1h 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"76029953dd1d132f1372c6b5b004370d9b4112d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 04:20:36.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"8d9f7ed22150396e95b5ae8201a0e458867d5bee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 06:05:46.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c84c0d62ba1449acaddcde2ac6a543ec : SUCCESS in 6m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d34c29a42a2a4d1a98da6c3439a953d8 : SUCCESS in 6m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3c307301c214d0791e35e1629e0468a : SUCCESS in 5m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/286f6087f5cc47debbdc02563a2a6559 : SUCCESS in 23m 49s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4eeb613e52a54afe86162291a820acc1 : SUCCESS in 17m 36s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3d2de9d67c64438b43e135bbc0a1628 : FAILURE in 45m 29s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e93b738081584bacaa1659f91398d374 : SUCCESS in 1h 11m 25s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8fe6037f60ca45a098740b5e992f08f5 : SUCCESS in 30m 54s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8e699abb79b946b9b442b36b9fbb48ec : SUCCESS in 1h 42m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c52cd37c88ef453d9eba59d4a4d2b0ed : SUCCESS in 3m 26s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e0829a40d44d4b29871b9a3f11e426c2 : SUCCESS in 1h 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"6904cdce20e695a976e5a9ac18bc75a3c55fe790","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 06:05:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"b92555de70ac9a4eb2d62bc02d7bd6141cb0fa51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 06:08:35.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3448c482c2734969a6c0944f46d1cda7 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"cf1dd1ea62020f62bb151ffc3d65bdfd8f22fc25","revisions":{"48d35c4ebeeb2f10e2e593b299610feba59a3a85":{"kind":"REWORK","_number":1,"created":"2019-11-15 05:00:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/1"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 05:00:00.000000000","tz":540},"subject":"WIP: Add support to configure swift-recon-cron job","message":"WIP: Add support to configure swift-recon-cron job\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/48d35c4ebeeb2f10e2e593b299610feba59a3a85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/48d35c4ebeeb2f10e2e593b299610feba59a3a85"}]},"branch":"refs/heads/master"},"1a3493d51324cc058196cb5ea7783fe1941be27d":{"kind":"REWORK","_number":2,"created":"2019-11-15 07:04:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/2"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 07:04:39.000000000","tz":540},"subject":"WIP: Add support to configure swift-recon-cron job","message":"WIP: Add support to configure swift-recon-cron job\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1a3493d51324cc058196cb5ea7783fe1941be27d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1a3493d51324cc058196cb5ea7783fe1941be27d"}]},"branch":"refs/heads/master"},"823571685502f567af6b8decd2d9bb2e8009c7e8":{"kind":"REWORK","_number":3,"created":"2019-11-18 02:28:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/3"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-18 02:28:30.000000000","tz":540},"subject":"WIP: Add support to configure swift-recon-cron job","message":"WIP: Add support to configure swift-recon-cron job\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/823571685502f567af6b8decd2d9bb2e8009c7e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/823571685502f567af6b8decd2d9bb2e8009c7e8"}]},"branch":"refs/heads/master"},"f06a0fffecf2274b5a12093371cb229d6f384c73":{"kind":"REWORK","_number":4,"created":"2019-11-18 02:30:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/4"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-18 02:30:16.000000000","tz":540},"subject":"WIP: Add support to configure swift-recon-cron job","message":"WIP: Add support to configure swift-recon-cron job\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f06a0fffecf2274b5a12093371cb229d6f384c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f06a0fffecf2274b5a12093371cb229d6f384c73"}]},"branch":"refs/heads/master"},"cd3685e23f6fe9c143f4166646c5f36f8f3b60ab":{"kind":"REWORK","_number":5,"created":"2019-11-18 02:55:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/5"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-18 02:55:46.000000000","tz":540},"subject":"WIP: Add support to configure swift-recon-cron job","message":"WIP: Add support to configure swift-recon-cron job\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cd3685e23f6fe9c143f4166646c5f36f8f3b60ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cd3685e23f6fe9c143f4166646c5f36f8f3b60ab"}]},"branch":"refs/heads/master"},"02e36dacbb1bb982dd449c436b8fb414c44e6990":{"kind":"REWORK","_number":6,"created":"2019-11-18 04:11:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/6"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-18 04:11:06.000000000","tz":540},"subject":"WIP: Add support to configure swift-recon-cron job","message":"WIP: Add support to configure swift-recon-cron job\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/02e36dacbb1bb982dd449c436b8fb414c44e6990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/02e36dacbb1bb982dd449c436b8fb414c44e6990"}]},"branch":"refs/heads/master"},"9fef093a3076244e2237f7b13b9651f24bdaf156":{"kind":"REWORK","_number":7,"created":"2019-11-18 05:15:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/7"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-18 05:15:46.000000000","tz":540},"subject":"WIP: Add support to configure swift-recon-cron job","message":"WIP: Add support to configure swift-recon-cron job\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9fef093a3076244e2237f7b13b9651f24bdaf156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9fef093a3076244e2237f7b13b9651f24bdaf156"}]},"branch":"refs/heads/master"},"a56b70a464db0985bb8776ccd0be0bbe6c08bdc5":{"kind":"REWORK","_number":8,"created":"2019-11-18 05:56:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/8"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-18 05:56:22.000000000","tz":540},"subject":"WIP: Add support to configure swift-recon-cron job","message":"WIP: Add support to configure swift-recon-cron job\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a56b70a464db0985bb8776ccd0be0bbe6c08bdc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a56b70a464db0985bb8776ccd0be0bbe6c08bdc5"}]},"branch":"refs/heads/master"},"7aad5549232d2a440959743cda4a6b7a863971c4":{"kind":"REWORK","_number":9,"created":"2019-11-18 07:57:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/9"}}},"commit":{"parents":[{"commit":"57503cb95a178dfaa15b9031062063f6196866b1","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/57503cb95a178dfaa15b9031062063f6196866b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-18 07:57:42.000000000","tz":540},"subject":"WIP: Add support to configure swift-recon-cron job","message":"WIP: Add support to configure swift-recon-cron job\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7aad5549232d2a440959743cda4a6b7a863971c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7aad5549232d2a440959743cda4a6b7a863971c4"}]},"branch":"refs/heads/master"},"dabfd9a0d934d6423f5a7be2e7ad5f1f2206dd41":{"kind":"REWORK","_number":10,"created":"2020-01-30 13:39:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/10"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-01-30 13:39:13.000000000","tz":540},"subject":"WIP: Add support to configure swift-recon-cron job","message":"WIP: Add support to configure swift-recon-cron job\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dabfd9a0d934d6423f5a7be2e7ad5f1f2206dd41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/dabfd9a0d934d6423f5a7be2e7ad5f1f2206dd41"}]},"branch":"refs/heads/master"},"d4b762bdc6618ff9a1b08c7def2c0b5029029883":{"kind":"REWORK","_number":11,"created":"2020-01-30 13:48:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/11"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-01-30 13:47:30.000000000","tz":540},"subject":"Add support for swift-recon-cron job","message":"Add support for swift-recon-cron job\n\nAdd support to configure swift-recon-cron job, which is\nrequired to have object async pendings shown in recon API.\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d4b762bdc6618ff9a1b08c7def2c0b5029029883"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d4b762bdc6618ff9a1b08c7def2c0b5029029883"}]},"branch":"refs/heads/master"},"35c385da82202e0ec8dd7ac0a1b55d8f356b343a":{"kind":"REWORK","_number":12,"created":"2020-01-30 14:12:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/12"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-01-30 14:12:22.000000000","tz":540},"subject":"Add support for swift-recon-cron job","message":"Add support for swift-recon-cron job\n\nAdd support to configure swift-recon-cron job, which is\nrequired to have object async pendings shown in recon API.\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/35c385da82202e0ec8dd7ac0a1b55d8f356b343a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/35c385da82202e0ec8dd7ac0a1b55d8f356b343a"}]},"branch":"refs/heads/master"},"c01ef10172da97f26903ce1440a0cb2be2600051":{"kind":"REWORK","_number":13,"created":"2020-01-30 23:37:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/13"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-01-30 23:37:32.000000000","tz":540},"subject":"Add support for swift-recon-cron job","message":"Add support for swift-recon-cron job\n\nAdd support to configure swift-recon-cron job, which is\nrequired to have object async pendings shown in recon API.\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c01ef10172da97f26903ce1440a0cb2be2600051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c01ef10172da97f26903ce1440a0cb2be2600051"}]},"branch":"refs/heads/master"},"8ec770bb50e582566864a119f066f205a730ff04":{"kind":"REWORK","_number":14,"created":"2020-01-31 02:05:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/14"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-01-31 02:05:13.000000000","tz":540},"subject":"Add support for swift-recon-cron job","message":"Add support for swift-recon-cron job\n\nAdd support to configure swift-recon-cron job, which is\nrequired to have object async pendings shown in recon API.\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8ec770bb50e582566864a119f066f205a730ff04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8ec770bb50e582566864a119f066f205a730ff04"}]},"branch":"refs/heads/master"},"3ea03f59cf920f0d1e6d9e11f6459d9f510b224c":{"kind":"REWORK","_number":15,"created":"2020-01-31 03:14:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/15"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-01-31 03:13:42.000000000","tz":540},"subject":"Add support for swift-recon-cron job","message":"Add support for swift-recon-cron job\n\nAdd support to configure swift-recon-cron job, which is\nrequired to have object async pendings shown in recon API.\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3ea03f59cf920f0d1e6d9e11f6459d9f510b224c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3ea03f59cf920f0d1e6d9e11f6459d9f510b224c"}]},"branch":"refs/heads/master"},"23b4bfb9ac3a2c9fc3a22ba7cc3b54baf2cf690b":{"kind":"REWORK","_number":16,"created":"2020-01-31 04:34:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/16"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-01-31 04:34:04.000000000","tz":540},"subject":"Add support for swift-recon-cron job","message":"Add support for swift-recon-cron job\n\nAdd support to configure swift-recon-cron job, which is\nrequired to have object async pendings shown in recon API.\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/23b4bfb9ac3a2c9fc3a22ba7cc3b54baf2cf690b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/23b4bfb9ac3a2c9fc3a22ba7cc3b54baf2cf690b"}]},"branch":"refs/heads/master"},"cf1dd1ea62020f62bb151ffc3d65bdfd8f22fc25":{"kind":"REWORK","_number":17,"created":"2020-01-31 07:57:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/694451/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/51/694451/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/51/694451/17"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2019-11-15 04:59:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-01-31 07:57:23.000000000","tz":540},"subject":"Add support for swift-recon-cron job","message":"Add support for swift-recon-cron job\n\nAdd support to configure swift-recon-cron job, which is\nrequired to have object async pendings shown in recon API.\n\nChange-Id: Iae3705db38ebaddf92ee6e80c20c1011f8f66706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cf1dd1ea62020f62bb151ffc3d65bdfd8f22fc25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cf1dd1ea62020f62bb151ffc3d65bdfd8f22fc25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
