)]}'
{"id":"openstack%2Fpuppet-swift~707168","triplet_id":"openstack%2Fpuppet-swift~master~I977ebb769eb79c715e0b19e97b9087bcbf578809","project":"openstack/puppet-swift","branch":"master","topic":"swift-ks-region","hashtags":[],"change_id":"I977ebb769eb79c715e0b19e97b9087bcbf578809","subject":"Add region parameters for keystone authentication","status":"MERGED","created":"2020-02-11 15:03:47.000000000","updated":"2020-02-19 22:45:31.000000000","submitted":"2020-02-19 22:42:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707168-1582152165451-25e6796c","meta_rev_id":"08806ba9791a1cc732bbe131be5a55f0b5801a8a","_number":707168,"virtual_id_number":707168,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 22:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-19 20:35:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-02-18 00:45:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-02-19 20:35:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 00:45:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-19 20:35:36.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-19 22:42:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e6d1d69d2fc2312f27b4d44653ff552f10e78cc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-11 15:03:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f29074d7848765653b859fc47eeedc9029058afc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-11 15:17:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ef9df8f2a099af392b9f5ccef14dd5f8779c9da7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-11 15:23:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b40274940f19f6228fb0f60cfb5442223971771d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-11 15:25:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4a19b3e27a6995d58b0e47dcc079547f8601154b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 19:29:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0ba9e6c5ca0427483e57be72e21226f : RETRY_LIMIT in 2m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9986be16854f4118b1937a3239e87ead : SUCCESS in 5m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c05746aab684d9b9bb246286faccba8 : RETRY_LIMIT in 3m 18s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7f3e5848b2a04cab9a35546ff6f3b9e5 : SUCCESS in 28m 47s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a49e98a2eef410cb4ec4c2ad80d409a : SUCCESS in 15m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c473cd303fa417ca6ccb19f3fad41e3 : SUCCESS in 17m 54s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00310f7eff2849979403f0610a4ee93c : SUCCESS in 53m 14s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8413b61b75004661ab5bbeb7c10e3342 : SUCCESS in 1h 08m 36s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/52dcb001003c44c0b80daa03f6bfcbb9 : SUCCESS in 1h 03m 11s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5fadfe17ec4843e182e6997eaa9f2636 : SUCCESS in 31m 25s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2cff99a47bba43b59cbe1d298cdbc598 : FAILURE in 25m 05s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7c0a30ed0de341b8b9f4b4b98f0a3ef1 : FAILURE in 11m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca9fca7bd2c74ddeb647ff0f8a171d78 : SUCCESS in 6m 28s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ef85077629f44e3aa2e0f39388628cb3 : FAILURE in 11m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9ce5edf22d50cd76292cc8176ec9ca53fd8bae8b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-11 23:04:13.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"83ed702260f81d817946200527b756807ec94beb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 00:18:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc0bc52ac2864ffbb6cc3cd72afbcc28 : SUCCESS in 7m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8b7da68d26b4571b2c6816dc77aeb40 : SUCCESS in 5m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22190511424a4fe4b917395365ce7e07 : SUCCESS in 8m 07s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/72be51086518421db663c2b94ee8c44d : SUCCESS in 27m 50s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4813c6c99b034bc8b1d74745ed992267 : SUCCESS in 17m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/769e8c445408468cb33df8c300c34704 : SUCCESS in 16m 40s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5d24bb18aac411eadcdb63bc20e2250 : SUCCESS in 51m 29s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/a6251149c11044b1be42bfd0da62995a : SUCCESS in 1h 07m 41s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/537a50f49d5b49be8208d9a40563c7a3 : SUCCESS in 1h 06m 11s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/294ab52bffde43fe9557072a840775b8 : SUCCESS in 33m 15s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8fbc63a0ca374890b6e55d42d4a1a732 : FAILURE in 32m 13s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e4ead0869fa44f838c46e335f662a741 : FAILURE in 11m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7398dbccfee46e58073da83f0d6bb49 : SUCCESS in 4m 39s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6c5d27c8cdfd438cad68b2e4d358e6e7 : FAILURE in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"13dd2ff8579636af0566ebe1f4431da887cec0b0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-12 02:47:26.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f3223443baf535c8bbfb2c22bd406c873361a435","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-14 02:32:17.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"e318ad07d93fc634c500a533c3ff5934660786ac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-14 02:32:57.000000000","message":"Topic changed from ceilometer-auth_uri to swift-ks-region","accounts_in_message":[],"_revision_number":6},{"id":"ac7c395c5994f762994c78709a848d53b759272e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 04:14:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9da17f6364254081a516cd35ad6c0342 : SUCCESS in 6m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/381ceaf34731413aae3fabff0c89595a : SUCCESS in 6m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06e11101ebf8498289e0e134ea58a751 : SUCCESS in 6m 02s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4411858dd01a45eb84e39d2bab0b869b : SUCCESS in 19m 50s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc2a6525c7df4ce0a6fe72fc9ebe6609 : SUCCESS in 14m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca5f2f9a02e34c30959a7f00a467ab6c : SUCCESS in 14m 39s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83b3da8e05114c4bb942c2825a3bd808 : SUCCESS in 49m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/3078974dee2d4910b31c67e1b2831996 : SUCCESS in 1h 00m 32s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c51353667268436994b003c2f357c75e : SUCCESS in 1h 02m 45s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/240351acb46e48c8845b4489ad4f21b0 : SUCCESS in 27m 55s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a1c9e8698704186b19fe971ee8b0a20 : FAILURE in 24m 44s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6a05f44da4cf4fe1b4459722e74e0090 : SUCCESS in 1h 41m 08s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73184de05db04d8a9f2e022051b1c67a : SUCCESS in 5m 01s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/76a309c44b7b4981a9c870e9ed98eda7 : SUCCESS in 1h 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"67666208b580e5a29f1248dda27991bf2ce4c42c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-17 20:35:02.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0309443f4cc0366c2f32e06f370084bea75d4873","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 22:48:27.000000000","message":"Patch Set 6: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"7d02144c526738aa490702b4d735c2598bfc7fd2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-17 23:02:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"740fb5a74f2e5f09bc1d7d7bc1b3d5d911c50fad","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-17 23:05:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"bcd1b863d0c6ddd9d261e0e9e98f227b7a059680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 00:43:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5be2309369ef42668bc081fe845d7550 : SUCCESS in 5m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a27165cfa3794667bbcec103dbde2af0 : SUCCESS in 5m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c616cd2c192748a49a6bfbeaf1e2b5c6 : SUCCESS in 6m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/a2148da6795943d589fb3e6ba8929a91 : SUCCESS in 25m 47s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc485b2f878740689487697b7b5ca0f3 : SUCCESS in 14m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/036d8d4f661f4ee5b5915eb5b40b25e7 : SUCCESS in 17m 05s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7ce4ad06e964d4c88700f106687f63a : SUCCESS in 54m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/433867286c874e0393e086d1af8943b9 : SUCCESS in 1h 10m 43s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a0b75705fc264a0295cf473e95ba7c3c : FAILURE in 1h 04m 09s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d8a001897f53461d958fa1bffbbf1de4 : SUCCESS in 36m 33s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6fae9982557452dae4347c8fbfbedc5 : FAILURE in 29m 12s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/bda1a4b947ee4f0f96652109f9384869 : SUCCESS in 1h 37m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e42f9d8056944ba788a6d95532113124 : SUCCESS in 4m 58s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f9e0b772cc3741758a17492c5579a1ca : SUCCESS in 1h 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"92dc1148be6bad691a9be06d347c9905dbca9c8c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-18 00:45:14.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"9785c96df43e1a9daefc275c96b32398e9937995","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-19 20:35:36.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"826bbd8fc3442c5711980cf12ae392d760608bd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 20:35:49.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2ff5aca73411d6bf9ccf7f2b6f4f48f25c58741e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 22:42:45.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/432cf749fd714273b7f51f1d2096fa64 : SUCCESS in 8m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b29d552f5ad40368d894e45ac9e2a02 : SUCCESS in 7m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/318bed8d5d924f159f15f23eac00cccb : SUCCESS in 7m 18s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/365259e294fd4419bc2feb5d15a4ee54 : SUCCESS in 28m 35s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b182af003e74dcc9a5ccc18bb09711d : SUCCESS in 20m 54s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c45a47f1a9604950af744bb75f79665a : SUCCESS in 52m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d0ae0e27c547479cbb16f169f06e2056 : SUCCESS in 1h 07m 16s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f4e4dcc25ee747daacf1613c066078e4 : SUCCESS in 33m 18s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7f7dd7a4f0e84ed29c8ea422c43fd140 : SUCCESS in 2h 01m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c9fe6c94ec742f7b71e6cdcd3d9c00a : SUCCESS in 5m 18s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0bc5d3d90ff24d5489838b9cb0f9a812 : SUCCESS in 1h 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"40bc677a6c1bea1fa16ab8185104550c428f9091","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 22:42:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"08806ba9791a1cc732bbe131be5a55f0b5801a8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 22:45:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/239f4e0a3abc4cd481485d89e8610c1a : SUCCESS in 2m 16s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"9344031dafb72336ce47c7922651581493f0ea95","revisions":{"f51ff7b72bece7c590e7d5106332bb46c4017e8c":{"kind":"REWORK","_number":1,"created":"2020-02-11 15:03:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/707168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/68/707168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/1"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 14:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 15:03:24.000000000","tz":540},"subject":"Add region parameter for authtoken middlware","message":"Add region parameter for authtoken middlware\n\nAdd swift::proxy::authtoken::region_name to specify keystone region\nwhere we expect all keystone resources like system user are created.\n\nChange-Id: I977ebb769eb79c715e0b19e97b9087bcbf578809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f51ff7b72bece7c590e7d5106332bb46c4017e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f51ff7b72bece7c590e7d5106332bb46c4017e8c"}]},"branch":"refs/heads/master"},"82f04080a3f5bd3993283439605c56a7ab115ac2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-11 15:17:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/707168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/68/707168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/2"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 14:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 15:16:58.000000000","tz":540},"subject":"Add region parameter for authtoken middlware","message":"Add region parameter for authtoken middlware\n\nAdd swift::proxy::authtoken::region_name to specify the keystone region\nwhere we expect all keystone resources like system user are created.\n\nChange-Id: I977ebb769eb79c715e0b19e97b9087bcbf578809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/82f04080a3f5bd3993283439605c56a7ab115ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/82f04080a3f5bd3993283439605c56a7ab115ac2"}]},"branch":"refs/heads/master"},"bfa91719c13b17299d51f1c0f3ec3370c6fd17cc":{"kind":"REWORK","_number":3,"created":"2020-02-11 15:23:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/707168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/68/707168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/3"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 14:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 15:23:47.000000000","tz":540},"subject":"Add region parameters for identity resources","message":"Add region parameters for identity resources\n\nAdd region_name parameters to authtoken middlleware and ceilometer\nmiddleware, so that we can specify the keystone region where we expect\nkeystone resources like swift user are created.\n\nChange-Id: I977ebb769eb79c715e0b19e97b9087bcbf578809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bfa91719c13b17299d51f1c0f3ec3370c6fd17cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/bfa91719c13b17299d51f1c0f3ec3370c6fd17cc"}]},"branch":"refs/heads/master"},"ec8ff24faeb0f05ae138ffaa3751f388f71bb994":{"kind":"REWORK","_number":4,"created":"2020-02-11 15:25:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/707168/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/68/707168/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/4"}}},"commit":{"parents":[{"commit":"12b62e289c35af587d6a8d688d5103eb05f35d98","subject":"Merge \"Allow users to change container-replicator interval value.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12b62e289c35af587d6a8d688d5103eb05f35d98"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 14:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 15:25:43.000000000","tz":540},"subject":"Add region parameters for keystone authentication","message":"Add region parameters for keystone authentication\n\nAdd region_name parameters for keystone authentication in authtoken\nmiddlleware and ceilometer middleware, so that we can specify the\nkeystone region where we expect keystone resources like swift user\nare created.\n\nChange-Id: I977ebb769eb79c715e0b19e97b9087bcbf578809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ec8ff24faeb0f05ae138ffaa3751f388f71bb994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ec8ff24faeb0f05ae138ffaa3751f388f71bb994"}]},"branch":"refs/heads/master"},"3397d2af98c9c4746be3b4fd7c81d0d8d4bc25c2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-02-11 23:04:13.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/68/707168/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/68/707168/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/5"}}},"commit":{"parents":[{"commit":"560c3d66ef35d1964b3215744740060bd0a09682","subject":"Merge \"Add support for swift-recon-cron job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/560c3d66ef35d1964b3215744740060bd0a09682"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 14:58:58.000000000","tz":540},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-02-11 23:04:13.000000000","tz":0},"subject":"Add region parameters for keystone authentication","message":"Add region parameters for keystone authentication\n\nAdd region_name parameters for keystone authentication in authtoken\nmiddlleware and ceilometer middleware, so that we can specify the\nkeystone region where we expect keystone resources like swift user\nare created.\n\nChange-Id: I977ebb769eb79c715e0b19e97b9087bcbf578809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3397d2af98c9c4746be3b4fd7c81d0d8d4bc25c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3397d2af98c9c4746be3b4fd7c81d0d8d4bc25c2"}]},"branch":"refs/heads/master"},"39801e366dc4fa515a4d5c7350b76f339aebd90f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-02-14 02:32:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/707168/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/68/707168/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/6"}}},"commit":{"parents":[{"commit":"48a8fc8ff8d1c024dd4fafcb63fe9c8f41edd61c","subject":"Prepare Ussuri M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/48a8fc8ff8d1c024dd4fafcb63fe9c8f41edd61c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 14:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-14 02:31:12.000000000","tz":540},"subject":"Add region parameters for keystone authentication","message":"Add region parameters for keystone authentication\n\nAdd region_name parameters for keystone authentication in authtoken\nmiddlleware and ceilometer middleware, so that we can specify the\nkeystone region where we expect keystone resources like swift user\nare created.\n\nChange-Id: I977ebb769eb79c715e0b19e97b9087bcbf578809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/39801e366dc4fa515a4d5c7350b76f339aebd90f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/39801e366dc4fa515a4d5c7350b76f339aebd90f"}]},"branch":"refs/heads/master"},"3b23a1384f0f030f95bf3af3ee73fb2c7a27bbea":{"kind":"REWORK","_number":7,"created":"2020-02-17 23:02:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/707168/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/68/707168/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/7"}}},"commit":{"parents":[{"commit":"b6b458cd59bde37635a1c146e6c76a62505be4e3","subject":"Add support for service token parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b6b458cd59bde37635a1c146e6c76a62505be4e3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 14:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-17 23:00:58.000000000","tz":540},"subject":"Add region parameters for keystone authentication","message":"Add region parameters for keystone authentication\n\nAdd region_name parameters for keystone authentication in authtoken\nmiddlleware and ceilometer middleware, so that we can specify the\nkeystone region where we expect keystone resources like swift user\nare created.\n\nChange-Id: I977ebb769eb79c715e0b19e97b9087bcbf578809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3b23a1384f0f030f95bf3af3ee73fb2c7a27bbea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3b23a1384f0f030f95bf3af3ee73fb2c7a27bbea"}]},"branch":"refs/heads/master"},"9344031dafb72336ce47c7922651581493f0ea95":{"kind":"REWORK","_number":8,"created":"2020-02-17 23:05:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/707168/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/68/707168/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/68/707168/8"}}},"commit":{"parents":[{"commit":"b6b458cd59bde37635a1c146e6c76a62505be4e3","subject":"Add support for service token parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b6b458cd59bde37635a1c146e6c76a62505be4e3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-11 14:58:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-17 23:04:54.000000000","tz":540},"subject":"Add region parameters for keystone authentication","message":"Add region parameters for keystone authentication\n\nAdd region_name parameters for keystone authentication in authtoken\nmiddlleware and ceilometer middleware, so that we can specify the\nkeystone region where we expect keystone resources like swift user\nare created.\n\nChange-Id: I977ebb769eb79c715e0b19e97b9087bcbf578809\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9344031dafb72336ce47c7922651581493f0ea95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9344031dafb72336ce47c7922651581493f0ea95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
