)]}'
{"id":"openstack%2Fpuppet-swift~707748","triplet_id":"openstack%2Fpuppet-swift~master~Id8738a52d0ad0c6aaa4cbab70a0b3461870017dd","project":"openstack/puppet-swift","branch":"master","topic":"swift-ceilo-auth_uri","hashtags":[],"change_id":"Id8738a52d0ad0c6aaa4cbab70a0b3461870017dd","subject":"Deprecate swift::proxy::ceilometer::auth_uri","status":"MERGED","created":"2020-02-14 02:32:17.000000000","updated":"2020-02-23 18:34:17.000000000","submitted":"2020-02-23 18:32:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707748-1582482775818-4f1731f4","meta_rev_id":"315a97a25f730212ddd04692bd51ff745804a251","_number":707748,"virtual_id_number":707748,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-23 18:32:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-17 23:06:29.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-02-17 23:06:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-02-23 16:30:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-16 23:27:26.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-17 20:35:06.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-23 16:30:44.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-02-23 18:32:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3c13e6b8d14aeef0cce23a56cec99e39927c3e0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-14 02:32:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45dd21dfa0d3f75c0bc244857af2d7be7b94d917","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-14 02:35:04.000000000","message":"Patch Set 1:\n\nSee https://github.com/openstack/ceilometermiddleware/blob/master/ceilometermiddleware/swift.py#L48-L57 to find currently available parameters","accounts_in_message":[],"_revision_number":1},{"id":"103432475a5cc0a7d4a6e4d216b5a9e49cd181f1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-14 02:35:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"402a689ba10ff0cbf918d9eda54c61d25bbe18f7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-14 02:39:03.000000000","message":"Topic changed from ceilometer-auth_uri to swift-ceilo-auth_uri","accounts_in_message":[],"_revision_number":2},{"id":"863b4013e34e6d3918a2eeed5da95a0c962f4a4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 04:02:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a9ad5661af1b4d7fbfea6e69fc3e05e1 : SUCCESS in 5m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bbb8f452275477f8e310f43a0489964 : SUCCESS in 7m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1c251bc743b4f8fb97a8cbe3a6d9593 : SUCCESS in 6m 25s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c3cb77c979e94444a7924771c3cf36a9 : SUCCESS in 19m 59s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/462443f21ddd41fb847c779f38fce924 : SUCCESS in 16m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/075ad7a4bde947d58403ddb69fdc87f7 : SUCCESS in 14m 45s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16591716a33c429ca09a0b84ff70b229 : SUCCESS in 49m 06s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0c14212fd03b4239ac26bc7ba6280a3f : SUCCESS in 57m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/067d47163cff4455b07aa2139df385b2 : FAILURE in 1h 14m 01s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6905eb99235d4f9692e1e57e2c5fed77 : SUCCESS in 25m 41s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fde945034e0e4eaa8af9b1ba630e656a : FAILURE in 28m 08s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5ab83775adf749e5941ed85f9a7097b4 : FAILURE in 1h 10m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd8aee797f86463faff31b6c6649ba4c : SUCCESS in 4m 16s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/234549271ef7400f8a9ffa6138b6b9fb : SUCCESS in 1h 22m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f8483228f44e4d06febe3337fe009529a67e31d3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-16 23:27:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9c1cbe29319d72978f2f49eee1a2d10500cf583c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-17 20:35:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1a5e07e001679534b06bb4341c5439ac84afa236","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 22:48:27.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"8e9b2b1bbdd00a241874e2ed9acb53a9ea869301","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-17 23:02:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"bf3ea7c858f6526d5b7167d47a61bba1a226549c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-17 23:06:29.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"5da49fab469fd2d26bd2ee70dfb45f1da25b3890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 00:48:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76c85de76a914b79a78737a31305da4e : SUCCESS in 5m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c36caecb36444a9ab944f01e7cd60a6 : SUCCESS in 6m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39cd24188b56496597d911ff73f6ed6f : SUCCESS in 7m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/16e56db9f46e493292c0ab03cc0ab5a6 : SUCCESS in 22m 12s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/93477ba7e713463f9ef1823cb41aac56 : SUCCESS in 19m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9b11cd727f644a6b1a7969d6c3480d8 : SUCCESS in 16m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac6b756d1ac14e18883d583cc2900f7e : SUCCESS in 48m 42s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c537a38af2bb43928fbc9c1c1b82270f : SUCCESS in 1h 04m 18s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1ad8b1c41b9d4f25bb7d6866dca3e9a0 : FAILURE in 41m 41s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e622924301214a22bd480f43cea67547 : SUCCESS in 30m 51s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cccd18653d2340cd80da9e26b50cc432 : FAILURE in 27m 04s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/0982d13988c544d1ae61b8fba012735b : SUCCESS in 1h 39m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a74168e5d5f345bebe2f99b5f03ba82b : SUCCESS in 4m 59s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/546d190707ca46e69a71c437d362029a : SUCCESS in 1h 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"760d9d02d3969189831ca88b7bd97eabaeb07010","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-02-23 16:30:44.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"429a2d742804415ac25c789f9047805e8506519d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 16:30:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e7ec175cec435da3df7deed5038adbea2eae7456","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 18:32:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c6ad13a6ed98941ca7931d5df2a8a7a4f19f5ada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 18:32:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bcccb3fe24ff43518a1ef7fbc24e80a2 : SUCCESS in 6m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e195922ed3c44abd81573da4dfcc11db : SUCCESS in 6m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08c089c09a2342b8a45242ec2bccc41f : SUCCESS in 5m 38s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b4d3aedd13ee4f7a8d5e522ce774fef3 : SUCCESS in 18m 05s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da6f13c195fd4e0e8fac9c9350a1ddda : SUCCESS in 18m 05s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bf55db06afb49c3a35738f927664573 : SUCCESS in 49m 52s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8f83381a41c6459bb731d9a7c15a944f : SUCCESS in 1h 04m 36s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e5be4e6909a24f4eb84b5c35eba64345 : SUCCESS in 28m 50s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/7c37f0d236b84af49cfce5c64efeaf84 : SUCCESS in 1h 57m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a3ac16633c647f1912fce8a89413404 : SUCCESS in 4m 39s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6f5028d0367a4deca0471f3ac2526025 : SUCCESS in 1h 26m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"315a97a25f730212ddd04692bd51ff745804a251","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 18:34:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c07e5dacd67f4a6588e242cedf897f0c : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1e57622f25b76d1ad19e2310669252ece24c0b9a","revisions":{"79a513dae328fec46565a0f8bbf949fb2dd8805d":{"kind":"REWORK","_number":1,"created":"2020-02-14 02:32:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/707748/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/48/707748/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/1"}}},"commit":{"parents":[{"commit":"39801e366dc4fa515a4d5c7350b76f339aebd90f","subject":"Add region parameters for keystone authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/39801e366dc4fa515a4d5c7350b76f339aebd90f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-14 02:27:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-14 02:31:20.000000000","tz":540},"subject":"Drop swift::proxy::ceilometer::auth_uri","message":"Drop swift::proxy::ceilometer::auth_uri\n\nauth_uri parameter is not used in ceilometermiddleware, so deprecate\npuppet parameter according to it.\n\nChange-Id: Id8738a52d0ad0c6aaa4cbab70a0b3461870017dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/79a513dae328fec46565a0f8bbf949fb2dd8805d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/79a513dae328fec46565a0f8bbf949fb2dd8805d"}]},"branch":"refs/heads/master"},"5fe7731a61ff4f499686801fdb136bbb490dff66":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-14 02:35:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/707748/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/48/707748/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/2"}}},"commit":{"parents":[{"commit":"39801e366dc4fa515a4d5c7350b76f339aebd90f","subject":"Add region parameters for keystone authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/39801e366dc4fa515a4d5c7350b76f339aebd90f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-14 02:27:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-14 02:35:36.000000000","tz":540},"subject":"Deprecate swift::proxy::ceilometer::auth_uri","message":"Deprecate swift::proxy::ceilometer::auth_uri\n\nauth_uri parameter is not used in ceilometermiddleware, so deprecate\npuppet parameter according to it.\n\nChange-Id: Id8738a52d0ad0c6aaa4cbab70a0b3461870017dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5fe7731a61ff4f499686801fdb136bbb490dff66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5fe7731a61ff4f499686801fdb136bbb490dff66"}]},"branch":"refs/heads/master"},"93690c4a0dca8c1ac25b075c0a899e1a926c2e00":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-17 23:02:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/707748/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/48/707748/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/3"}}},"commit":{"parents":[{"commit":"3b23a1384f0f030f95bf3af3ee73fb2c7a27bbea","subject":"Add region parameters for keystone authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3b23a1384f0f030f95bf3af3ee73fb2c7a27bbea"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-14 02:27:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-17 23:00:58.000000000","tz":540},"subject":"Deprecate swift::proxy::ceilometer::auth_uri","message":"Deprecate swift::proxy::ceilometer::auth_uri\n\nauth_uri parameter is not used in ceilometermiddleware, so deprecate\npuppet parameter according to it.\n\nChange-Id: Id8738a52d0ad0c6aaa4cbab70a0b3461870017dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/93690c4a0dca8c1ac25b075c0a899e1a926c2e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/93690c4a0dca8c1ac25b075c0a899e1a926c2e00"}]},"branch":"refs/heads/master"},"1e57622f25b76d1ad19e2310669252ece24c0b9a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-02-17 23:06:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/707748/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/48/707748/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/48/707748/4"}}},"commit":{"parents":[{"commit":"9344031dafb72336ce47c7922651581493f0ea95","subject":"Add region parameters for keystone authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9344031dafb72336ce47c7922651581493f0ea95"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-14 02:27:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-17 23:06:04.000000000","tz":540},"subject":"Deprecate swift::proxy::ceilometer::auth_uri","message":"Deprecate swift::proxy::ceilometer::auth_uri\n\nauth_uri parameter is not used in ceilometermiddleware, so deprecate\npuppet parameter according to it.\n\nChange-Id: Id8738a52d0ad0c6aaa4cbab70a0b3461870017dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1e57622f25b76d1ad19e2310669252ece24c0b9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1e57622f25b76d1ad19e2310669252ece24c0b9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
