)]}'
{"id":"openstack%2Fpuppet-swift~708040","triplet_id":"openstack%2Fpuppet-swift~master~If60373cb44a62397678cee7aaefad3431a91f7e4","project":"openstack/puppet-swift","branch":"master","topic":"service_token","hashtags":[],"change_id":"If60373cb44a62397678cee7aaefad3431a91f7e4","subject":"Add support for service token parameters","status":"MERGED","created":"2020-02-16 22:39:28.000000000","updated":"2020-02-17 22:50:22.000000000","submitted":"2020-02-17 22:48:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708040-1581979706706-e503d632","meta_rev_id":"623849a46bc8c88b934fd22891b540bdca5a5960","_number":708040,"virtual_id_number":708040,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-17 22:48:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-17 14:19:02.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-02-17 20:34:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-02-17 20:34:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 14:19:02.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-17 20:34:56.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-02-17 22:48:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ec040f5aa7be914c951f08307c3898974dce9cb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-02-16 22:39:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f04f38d8c9afc098fc11fd10f7504bb444037e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 00:16:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b42e1b0c5ba945cda8eb26070d961588 : SUCCESS in 5m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/844cb0fbc9214dd0865fd4c623629469 : SUCCESS in 7m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e7e365e119d4a2e8be308ea057789a0 : SUCCESS in 5m 51s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/e5bf27ab4a594101ade728789e70ade9 : SUCCESS in 18m 16s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd8b1de46cdf454c98005665dd9adf6c : SUCCESS in 13m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6344a40507b542cebcd91c34e09475a5 : SUCCESS in 14m 55s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/343f27bdcd23431bad340b5886bd8eb8 : SUCCESS in 49m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/90838094308c40e3bf0c535165b0f266 : SUCCESS in 1h 02m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/70be10876c2841359407823906eef7fc : SUCCESS in 1h 06m 00s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3e7214467f384d9f8484f3c714f54729 : SUCCESS in 27m 52s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0114b436c9664571891c1f3b863ffad0 : FAILURE in 25m 28s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5b24c405dbeb4179abd0b871b02f52e3 : SUCCESS in 1h 36m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8967151c722c42909c7d70d871b26093 : SUCCESS in 4m 28s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/609aa2745ac543418bcfa6b02c6b3acc : SUCCESS in 1h 12m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a43b6a0ff7d0f7966fe7deecd7a359e35fe3cacd","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-17 14:19:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3387fd0ed6d2daaaa4ae060a9eda9574b8918172","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-02-17 20:34:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2844ea83d3a47ed58c422371c9c32cc9aa7034d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 20:35:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4391b96d5ff92f8390be183ebbbb8149948599c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 22:48:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"21d620542e5d42dfd015b17459ff86ee1f75dd48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 22:48:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c28c910b1ea47109a7e7e7df88e9135 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c3088d9d6254b7c9644a58340fc09ca : SUCCESS in 5m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b26caf959c741dc96e0474cab87453f : SUCCESS in 5m 48s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/aeb39dffd72c41dea095960db8bcafa9 : SUCCESS in 26m 27s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbeedadf4367421e9aeea68d63815bea : SUCCESS in 15m 15s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a57982842d624800a36726e91287aab5 : SUCCESS in 50m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ee95b1b1a1634470a21266b815999720 : SUCCESS in 1h 08m 37s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b747deeafd8d47f0b59a4955a5b557de : SUCCESS in 29m 33s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b172b3df6ccd46e0ae8f54812d465ccd : SUCCESS in 2h 08m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4e9123f73a441e68131efc7916b6262 : SUCCESS in 5m 17s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d5af3f3de2ad4a9896398ee9108578a9 : SUCCESS in 1h 21m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0f8d7054ba2c000993aba4b95e96d27bc76339e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 22:50:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af75ae82a6194dfab0058ed3def9e43f : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6b458cd59bde37635a1c146e6c76a62505be4e3","revisions":{"b6b458cd59bde37635a1c146e6c76a62505be4e3":{"kind":"REWORK","_number":1,"created":"2020-02-16 22:39:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/708040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/40/708040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/40/708040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/40/708040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/40/708040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/40/708040/1"}}},"commit":{"parents":[{"commit":"48a8fc8ff8d1c024dd4fafcb63fe9c8f41edd61c","subject":"Prepare Ussuri M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/48a8fc8ff8d1c024dd4fafcb63fe9c8f41edd61c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-16 22:38:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-02-16 22:38:33.000000000","tz":540},"subject":"Add support for service token parameters","message":"Add support for service token parameters\n\nAdd support for the parameters in authtoken middleware, which are\nrelated to service token feature.\n\nChange-Id: If60373cb44a62397678cee7aaefad3431a91f7e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b6b458cd59bde37635a1c146e6c76a62505be4e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b6b458cd59bde37635a1c146e6c76a62505be4e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
