)]}'
{"id":"openstack%2Fpuppet-swift~712635","triplet_id":"openstack%2Fpuppet-swift~master~Ib6105f3a06fe2e42864039356bf58dbe83749897","project":"openstack/puppet-swift","branch":"master","topic":"max_upload_part_num","hashtags":[],"change_id":"Ib6105f3a06fe2e42864039356bf58dbe83749897","subject":"Add max_upload_part_num param","status":"MERGED","created":"2020-03-12 10:10:12.000000000","updated":"2020-03-19 00:16:58.000000000","submitted":"2020-03-19 00:16:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712635-1584577018138-962e1dba","meta_rev_id":"416d015e49be0fa8cc722807965856487915c076","_number":712635,"virtual_id_number":712635,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 00:16:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-03-18 19:35:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-03-18 22:38:50.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-16 12:47:22.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-03-18 16:46:14.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-03-18 19:35:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-18 22:38:50.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-19 00:16:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cd4091cb0f54a25cc266161e518d2bd5afcbf20","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-03-12 10:10:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32e66c495829aa4efb9743d9ede6de1528d8f681","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 11:28:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/092e6e1c47a843189102f7032aeb1771 : SUCCESS in 7m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ec75c3667b247fb96ca139c6de194d2 : SUCCESS in 5m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a4f9b23316b4600bea28e99cd50a5ae : SUCCESS in 6m 08s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/abbb234ab276482bbc86f4f519a3d3e5 : SUCCESS in 22m 27s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d21d18a2fa6e417a99451f4e6d202d90 : SUCCESS in 20m 24s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/0d39b0e7094a48a09f72f4dd1a28e0b2 : SUCCESS in 24m 04s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d112871924b14e6a87daef622de00856 : SUCCESS in 15m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ed3d393a435493ab5d08591e6cf757f : SUCCESS in 16m 04s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b625e7506504894b95f09159aa4278d : SUCCESS in 58m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/2a81843273934129ba8f1cc27799734f : SUCCESS in 1h 03m 07s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/44aca273039043c1b511271c98e2a652 : SUCCESS in 1h 13m 08s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1ce64f8160284151a567606abef4cd35 : SUCCESS in 29m 43s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aabc40b924e415d830fbeceee93d9a2 : FAILURE in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34b17c2e49999ea041ad004ad977edb4907b4a0f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-12 15:23:52.000000000","message":"Patch Set 1: Code-Review+2\n\nwould be nice to cover with test","accounts_in_message":[],"_revision_number":1},{"id":"0cb40408ea0cce42a4dc0b5164ebb67337c5f64c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-03-16 12:47:22.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI won\u0027t put +1 atm because of a possible improvoment about the default value.\n\nI submitted a subsequent change to implement missing unit tests (for this parameter and one existing parameter), which can be merged independently as a follow-up.","accounts_in_message":[],"_revision_number":1},{"id":"f8ab037c16077885806b0e985365c4b73f1e60ca","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-03-18 16:47:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12c621eb1b015befa7f9d79387857c279bab145a","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-03-18 16:47:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9fcc5a7f19592cc5e524146ac02e23847498f3e9","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-03-18 16:50:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5c4c8f567acebeb0dd3962be5b845234819cb915","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-03-18 16:50:23.000000000","message":"Patch Set 1:\n\nAlex, I\u0027ve added a test, as you suggested.","accounts_in_message":[],"_revision_number":1},{"id":"8a773ff33de1408f8d927125b4ec7f5956617bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 18:41:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1abf4d2ffdb2460aac6be45f34dddabf : SUCCESS in 5m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/558389d5159c4ee698eb09781eef25d8 : SUCCESS in 7m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26fa372e9ac64a7e9d4a08510be8f0da : SUCCESS in 5m 14s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/91c9a498f5974694bc851626bbd94009 : SUCCESS in 25m 26s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b0dcc071d99f4b098859e6d0d1ae8511 : SUCCESS in 19m 49s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/89968fa06d7e49c6a4705a3688fa6c7d : SUCCESS in 18m 48s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f769ca8b86a4f89ad78927b60bb11a7 : SUCCESS in 22m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8ff8049df994f6f8058edb83b2970ed : SUCCESS in 23m 49s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4e0f671106e47edb8d6bc13d0f2b0a0 : SUCCESS in 1h 11m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/5e29e0efe47d448bbe54d94ac1b3e8da : SUCCESS in 1h 22m 21s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/29b5641e31e24f25b6479db743bdaeaa : FAILURE in 19m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5481b566a4044534af8f9f28a10460a4 : SUCCESS in 33m 44s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ad471a24ad046809870e35a0594b2fa : FAILURE in 12m 18s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5d969921d0604744bc851e716c3780f1 : SUCCESS in 1h 47m 41s","accounts_in_message":[],"_revision_number":3},{"id":"bd73db4e5f27a8a734bc54f78e33332c8e96fc20","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-18 19:35:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2cbbb2b8aad471a48dbcac642380418eaf423ba7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-18 22:38:50.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1d9fbea778a4317afc8de834eca76788be584270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 22:39:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"585f3a2df3d44c5d0ff8ea957d0be037f047942f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 00:16:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/844067f6cc2b4acab3c4481c00c12189 : SUCCESS in 4m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ebd2ada76904f19a52e508702b99089 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7b1004de96948e789915d6955ea003f : SUCCESS in 5m 29s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/110bdb70e24d42da9982376e0a2a5d03 : SUCCESS in 24m 09s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8109bc61209b40a98768bc8a1b7f9f4d : SUCCESS in 15m 32s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a81f823b17714834952d06a4900e9ee6 : SUCCESS in 51m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/da00e4623c8e4f0e84bcbe27221e6b7d : SUCCESS in 1h 05m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/091059620e8442779d4d645fb1506fa6 : SUCCESS in 26m 59s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2e9839da487e49d0a7054cd12ec63304 : SUCCESS in 1h 33m 25s","accounts_in_message":[],"_revision_number":3},{"id":"4ac230cc771c5140cbdcd4130a538429b35b89a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 00:16:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"294972d8fca90fe851cf68992fcd4c1457fd8886","revisions":{"98feae35cc80308b3100dfb428d0031d1d56c360":{"kind":"REWORK","_number":1,"created":"2020-03-12 10:10:12.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/35/712635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/35/712635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/1"}}},"commit":{"parents":[{"commit":"3b4e562faf4b098412ceb6fe25b660eea9b22cb7","subject":"Add object_server_mb_per_sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3b4e562faf4b098412ceb6fe25b660eea9b22cb7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-03-12 10:09:06.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-03-12 10:10:03.000000000","tz":60},"subject":"Add max_upload_part_num param","message":"Add max_upload_part_num param\n\nThe default in swift is 1000, though in S3, it is 10 000. Therefore,\nit is necessary to be able to configure it, so that a deployment\ncan be set with the same limits as S3.\n\nChange-Id: Ib6105f3a06fe2e42864039356bf58dbe83749897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/98feae35cc80308b3100dfb428d0031d1d56c360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/98feae35cc80308b3100dfb428d0031d1d56c360"}]},"branch":"refs/heads/master"},"9d7c232a50e648edb6a03bbe265642e889e8aa3a":{"kind":"REWORK","_number":2,"created":"2020-03-18 16:47:55.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/35/712635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/35/712635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/2"}}},"commit":{"parents":[{"commit":"3b4e562faf4b098412ceb6fe25b660eea9b22cb7","subject":"Add object_server_mb_per_sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3b4e562faf4b098412ceb6fe25b660eea9b22cb7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-03-12 10:09:06.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-03-18 16:47:42.000000000","tz":60},"subject":"Add max_upload_part_num param","message":"Add max_upload_part_num param\n\nThe default in swift is 1000, though in S3, it is 10 000. Therefore,\nit is necessary to be able to configure it, so that a deployment\ncan be set with the same limits as S3.\n\nChange-Id: Ib6105f3a06fe2e42864039356bf58dbe83749897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9d7c232a50e648edb6a03bbe265642e889e8aa3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9d7c232a50e648edb6a03bbe265642e889e8aa3a"}]},"branch":"refs/heads/master"},"294972d8fca90fe851cf68992fcd4c1457fd8886":{"kind":"REWORK","_number":3,"created":"2020-03-18 16:50:19.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/35/712635/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/35/712635/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/35/712635/3"}}},"commit":{"parents":[{"commit":"3b4e562faf4b098412ceb6fe25b660eea9b22cb7","subject":"Add object_server_mb_per_sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3b4e562faf4b098412ceb6fe25b660eea9b22cb7"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-03-12 10:09:06.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-03-18 16:50:05.000000000","tz":60},"subject":"Add max_upload_part_num param","message":"Add max_upload_part_num param\n\nThe default in swift is 1000, though in S3, it is 10 000. Therefore,\nit is necessary to be able to configure it, so that a deployment\ncan be set with the same limits as S3.\n\nChange-Id: Ib6105f3a06fe2e42864039356bf58dbe83749897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/294972d8fca90fe851cf68992fcd4c1457fd8886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/294972d8fca90fe851cf68992fcd4c1457fd8886"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
